)]}'
{"id":"openstack%2Frequirements~779146","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ic800d4f93cc324787fe22e27a25411e1565185d5","project":"openstack/requirements","branch":"stable/victoria","topic":"bug/1902696","hashtags":[],"change_id":"Ic800d4f93cc324787fe22e27a25411e1565185d5","subject":"Pin kombu and amqp requirements due to incompatibility","status":"ABANDONED","created":"2021-03-08 11:19:32.000000000","updated":"2022-08-02 09:25:03.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":761194,"cherry_pick_of_patch_set":5,"meta_rev_id":"88fddfac9f57be7b9b3d3f627f290e33b47e8028","_number":779146,"virtual_id_number":779146,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 13:39:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"reviewer_updates":[{"updated":"2021-03-08 11:19:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-03-08 11:40:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 13:39:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 15:48:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-03-08 17:49:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"91e7aef9201c2ae0128335edfbabaf3803c0fd38","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-08 11:19:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"8a8bd55cbf397afc9881ad1f80053724905a9332","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-08 11:22:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cc9568307333fe142b8de0dad877a2b1710b078f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-08 11:23:06.000000000","message":"Topic bug/1902696-stable/victoria removed","accounts_in_message":[],"_revision_number":2},{"id":"26b9a1a500720da8cb25808b754913122e4da252","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-08 11:23:18.000000000","message":"Topic set to bug/1902696","accounts_in_message":[],"_revision_number":2},{"id":"bc0b7abcb5808681df0595fcd82639923b721f47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:40:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6fac632ae07e4b668dbb7a3bb6737b57 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"022b125542e11ef4deb7ddc924d0a718708ab6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:39:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9475c6daa90342348eb3dbace019f415 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8f8493c381642d792f598e7cb06d1ef : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbcf683b0882411a82ad3bb7472e4a61 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dda0f301fc5a43f6935df241dade88d8 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/476d242894f14844901808813c0d884b : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/666a9c835bf64372853a764bfbe0f29d : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/507ad0a6356c486fb276ec6adf4c795f : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bce3929b9c26452a912d589765d6c5fb : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd2e56e3f9c94f0888f4f8e9fece4ec6 : SUCCESS in 9m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8f7ef68f80b44282b345aa2b96dd71bd : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c9bebdf81074327a0efe9527939d241 : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7935d4b92d8464c8cddfe476d843d0b : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/84bb349f6e97474d86e76f643b4914ed : SUCCESS in 15m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/47d5a60176584087a18c231f7eac4ae4 : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4553af55a7ce4a2199f4326b96dbde0f : SUCCESS in 14m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47c699fbfe73455fb31dace45b84b5bf : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f5afb7d60c7442dba656c768af85f48 : SUCCESS in 20m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1e59844da91046978387e252cf520957 : SUCCESS in 7m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d2a23143a024d4c93e796ee2bacc1ff : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b3c82a9aa1cd4728a7d864430874c331 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b48abda30d3848e4ad60f4c11f61ec38 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6a9caf50bde49ec9014919c168882c0 : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ec3ee0272ab0f4eeb1352f592afee07e3339391b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-08 15:48:05.000000000","message":"Patch Set 2:\n\nneed to remove the wip flag","accounts_in_message":[],"_revision_number":2},{"id":"b890cd0050b92ce4d29a69354550a68027cf5777","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 17:49:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88fddfac9f57be7b9b3d3f627f290e33b47e8028","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:25:03.000000000","message":"Abandoned\n\nThis is a very old WIP change.  Please restore and update if still needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02299afcfa9005122a20031dc1d45d6302eb0dd5","revisions":{"e8c161e058bbd9bfcb3d956fe46b1c0062069bbf":{"kind":"REWORK","_number":1,"created":"2021-03-08 11:19:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/779146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/779146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/779146/1"}}},"commit":{"parents":[{"commit":"8a4d1b57752eca95c63947e387db77046948c260","subject":"update constraint for oslo.cache to new release 2.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a4d1b57752eca95c63947e387db77046948c260"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-08 11:19:32.000000000","tz":0},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nLater amqp 5.0.2 reintroduced the ca_certs parameter but kolla now fails\nwith\n\nValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED\n\nTo keep the resulting constraints consistent the vine constraint moved\nback to 1.3.0 from 5.0.0\n\nThis patch partially reverts I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8c161e058bbd9bfcb3d956fe46b1c0062069bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8c161e058bbd9bfcb3d956fe46b1c0062069bbf"}]},"branch":"refs/heads/stable/victoria"},"02299afcfa9005122a20031dc1d45d6302eb0dd5":{"kind":"REWORK","_number":2,"created":"2021-03-08 11:22:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/779146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/779146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/779146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/779146/2"}}},"commit":{"parents":[{"commit":"8a4d1b57752eca95c63947e387db77046948c260","subject":"update constraint for oslo.cache to new release 2.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a4d1b57752eca95c63947e387db77046948c260"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-08 11:21:47.000000000","tz":0},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nLater amqp 5.0.2 reintroduced the ca_certs parameter but kolla now fails\nwith\n\nValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED\n\nTo keep the resulting constraints consistent the vine constraint moved\nback to 1.3.0 from 5.0.0\n\nThis patch partially reverts I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02299afcfa9005122a20031dc1d45d6302eb0dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02299afcfa9005122a20031dc1d45d6302eb0dd5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
