)]}'
{"id":"openstack%2Frequirements~779187","triplet_id":"openstack%2Frequirements~master~I64b02ac852872b190be829b698315a47f91881a0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I64b02ac852872b190be829b698315a47f91881a0","subject":"update constraint for os-brick to new release 4.3.0","status":"MERGED","created":"2021-03-08 09:34:05.000000000","updated":"2021-03-08 20:50:14.000000000","submitted":"2021-03-08 20:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779187-new-release","meta_rev_id":"64d34d369e6bf15791449afde0628c15de290889","_number":779187,"virtual_id_number":779187,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 20:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-08 17:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-08 09:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-08 17:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 09:54:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-08 10:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 11:26:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 17:33:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bcecf692b417e05482ca4f7fb08757ed7488c1c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-08 09:34:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3aaff91fed8be2023aac031a3f2b1c927f530cbe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 09:54:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2fca0629b554822c11c089c9544b3765b98ab5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 10:07:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d2a423b72db4aadbe077abf26bf00b0 : FAILURE in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b7b25176151a2b45a00dc30644f9d6dc8e07f727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42720990b4f84f6caea1bb5a07000bf1 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3edfec724dc414690ca0f420f622c63 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16f1b521946f48a48a2ae3c9068256db : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a0c50a7e3164978b4bc3ed1fbe8414a : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1787289156546c3b9801e5eb3f0b178 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb55abeb7b694971af5a4dd5c4addc35 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51e84d6d01494da295b52f3ca9961b88 : SUCCESS in 6m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8cf2c2f4a9d476695349db93962b937 : SUCCESS in 11m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/20d2b5ea95434951a0cc508b6edbc9b1 : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e716a0f23ee54b0d81f7cb9b81a962ca : SUCCESS in 11m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0dc390952b224c5996ec42ccd28af6d0 : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab05be6cc291474f9e5b6e6d64d1b22a : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21096f44a387476c8d88826ab1403187 : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3d2c747121054b958b02ec24d77dff68 : SUCCESS in 16m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6eaf12d3638a45e09e9ae6a4790573f0 : SUCCESS in 20m 14s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/935b24e3ce27493698456d2bc6b9b156 : SUCCESS in 6m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae5f864cdb1d491b806f72217838e42f : SUCCESS in 27m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa910ea791594b4e8e95c152e7506da4 : SUCCESS in 13m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/865355cf9b4445a98dfd7b91bec857af : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b57bb38dd394d8080bf542185203a7c : SUCCESS in 17m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f7b0fd048ed14112aa934d4f70f24df8 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca3dfe72dd924c379d561364ad40c60f : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8c258130b3c486b95c07bf9f6a6bfa5 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1cb1780ec114ce8b36c7fcdb3d4b8df : SUCCESS in 58m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c98564bac7244e218470b00ba3ae200d : SUCCESS in 1h 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7e0121a2b0dea0ee0945aac8928769a001a5e89c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 17:33:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5e84f3a963c91b97e7b74d03f7eae23c9050e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad5cd644802c1fcc51cd3d7fd1be727fb602c682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:35:44.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fc5e906a773c7d1129da80e69488fc8c290cb6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:48:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c5617d488d4338b589fa2b16826a36 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77d30df3fbd450d933f359f9bebe22a : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b64f1782ae4425b9fdff3c749210ede : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e7bf04ae3f499ab2ccc9da814754b9 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ade2e6b2418a4c2296736e189c670291 : SUCCESS in 9m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b86ff453ea941a2b158f3f9fa93672d : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d9e1d4bdb4f4aaabab7101410f61366 : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d506dec51c2342879fa1a1dcbd904527 : SUCCESS in 12m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfadbd9810cd42798add2136688f9f70 : SUCCESS in 10m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9bc33fa543f14a118f37e1d39cb2ff3b : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f5dc495358154562898639f5dccfbb96 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ac14a5f8cbb41f5912962c97cc6e000 : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4d2e974118dc4c2dab469c3b01077067 : SUCCESS in 13m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e736ddda7a87431ea37a8dd32915777b : SUCCESS in 15m 43s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f8e37d7d220b44a1a0b25d1a74106976 : SUCCESS in 6m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc70c2dfe23e4a4580b4ef8a4704c963 : SUCCESS in 30m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d64d0a8dd6a4fc9be575b5420cd1b0b : SUCCESS in 15m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa5fb00fc95f44b1944f15596f35a741 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ad99f245cc844a5b6f7c91142a15bdd : SUCCESS in 18m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9ab22d6ef3434f9da67af0cc95c8cda9 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33ce1bc02d9240b284547f0a0589dd82 : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65ab541d0ce1479d83a39180af1589c3 : SUCCESS in 17m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3a5f7c41f8e404db3c1e5a32b92e5dc : SUCCESS in 48m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ad17c6513224db08037f92df8b0cf74 : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1fb5c0ff3c44984aa2be239f1c400ee7566ff520","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64d34d369e6bf15791449afde0628c15de290889","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:50:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db25559d8f834fc49f1efd56ea21c398 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95074a1ab55a7222ce9865d7709504f0316c559c","revisions":{"95074a1ab55a7222ce9865d7709504f0316c559c":{"kind":"REWORK","_number":1,"created":"2021-03-08 09:34:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/779187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/779187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/779187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/779187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/779187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/779187/1"}}},"commit":{"parents":[{"commit":"79ff19fa57044575ea93b57c29c0e0b46028989d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ff19fa57044575ea93b57c29c0e0b46028989d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 09:34:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 09:34:03.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.3.0","message":"update constraint for os-brick to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I3035e6c233bbd8f1448a898235a7a4ef9ff7f3b3\nmeta: release:Code-Review+1: Luigi Toscano \u003cltoscano@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I64b02ac852872b190be829b698315a47f91881a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95074a1ab55a7222ce9865d7709504f0316c559c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95074a1ab55a7222ce9865d7709504f0316c559c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
