)]}'
{"id":"openstack%2Frequirements~779227","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ic6cb2d67dfe7026f809e31975622bde957d00b23","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ic6cb2d67dfe7026f809e31975622bde957d00b23","subject":"update constraint for python-cinderclient to new release 7.2.1","status":"MERGED","created":"2021-03-08 11:54:48.000000000","updated":"2021-03-08 21:00:28.000000000","submitted":"2021-03-08 21:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779227-new-release","meta_rev_id":"9c5ace8ef373c855d03e3e30111fec6a8001889b","_number":779227,"virtual_id_number":779227,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 21:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-08 17:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-08 11:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-08 17:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 11:56:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-08 12:16:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 14:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 17:34:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"65687bc4f2dcd0079d48d34a7ef28a9b2dff6335","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-08 11:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5087491c0631bcb1b720597f0f5540ca2370fbfc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 11:56:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68d8c5a75d0595a99cb6fd004e1fa02489fb91bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 12:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37ec8bf2257c4edfa4c90211af84a929 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f32fb54557857e4b817dd99b8bcf24a65292710e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 14:09:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25ca727f7bf41189b623a6928266ef3 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e8fe450cda4d85b9bae3187497b2e1 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c83e23324245408986511122630bbe56 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/537b8a56ef594c5c891106af4e4a2743 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b1fd51ef9a246d5a57214750f50a86a : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea1aeddddb9e4365939e163e7a747e0f : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/655d5d0cbe734ac095923ce1f382a692 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/075babe015ab455694467ec1e25bf6e2 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/398ae2b1cc6646b495f25588a2265f95 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/50e4bc2d776e498abf07ff95866a9731 : SUCCESS in 15m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69bce9473b5f4c51996afbcee800f5b8 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e87dd890838d452798e4fe89610483e7 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d8b8825d6b2b4d828e4a929b2a40db6d : SUCCESS in 14m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f343370916e44705a8f3fa29ecc7cca6 : SUCCESS in 24m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/714cd68ae2fa4355a32bc30fd9e65b79 : SUCCESS in 14m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f77a4d720c01410f9121eac8cd2a118e : SUCCESS in 9m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba1a1ed78d9b431a8a264077c8968546 : SUCCESS in 22m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4fde04b7c544bf1b24989e16c9f5512 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70e08128af174b91a8a5ccd7bd473599 : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e329fdb5e461428aadcb5d1a0ddb8167 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c56e5a775c92473d9506864b02e59dc3 : SUCCESS in 1h 02m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f2f19f27db1437d8b5c24adbbc01d1c : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":1},{"id":"36edf1ea55a96cbfebd58c344ed27ba20a915f7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 17:34:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3faf3ca2524b6e22232f3391ef3ff9e67fc06af9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:40:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12185b150ab08674bce87f56f98eac056c4c1da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:00:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc07248b5e04ab58059a97d6dd7cb9f : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a938f560a2f44271b34e02c922ab2164 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d436a2b02564c37892c140c4a73405c : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d099cc0eaacc49e3a2a9f5e75a1df04a : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf7415f99234472b842d147e017bdc4b : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ee7f31e987c4bc79385a5c42f9e46bf : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8241fb663fc4a019bb4716f04914b90 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2d068ab11c441e2add0a2e6d20c791c : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6fe3abc4a7284f8b8cd6021bda9608a3 : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ef8f97bdae91467c868a6608b0ef504f : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/986cc6d017d7427fbf79b9849a486d22 : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e06036099a6a4209adfc7e3d17fc933f : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f88b984be40748cfaafe2722a433738e : SUCCESS in 12m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8df22c2fe4ae4962a772e752f89b96d3 : SUCCESS in 28m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/462fb600c63e4f9c826ca8bc5b620bf7 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6570a597acac49a9bf1576a03f9763c4 : SUCCESS in 7m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a3068ef980446ddbcf061184b05fac2 : SUCCESS in 21m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f2ef9d26d03040bf9fb8e199e60c1db7 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2401e4f0bb14ea598ea4c5045395ada : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/99884ad4f8634fef8da6f43d4bb77666 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76eb547cd48242c2880e30a546941b96 : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb5c76e00f2c406c83eeeb76ec8d7ec7 : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9c5ace8ef373c855d03e3e30111fec6a8001889b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:00:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2b41891370b11ff27ac83a576f33f4c137bb032","revisions":{"e2b41891370b11ff27ac83a576f33f4c137bb032":{"kind":"REWORK","_number":1,"created":"2021-03-08 11:54:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/779227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/779227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/779227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/779227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/779227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/779227/1"}}},"commit":{"parents":[{"commit":"8a4d1b57752eca95c63947e387db77046948c260","subject":"update constraint for oslo.cache to new release 2.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a4d1b57752eca95c63947e387db77046948c260"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 11:54:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 11:54:45.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 7.2.1","message":"update constraint for python-cinderclient to new release 7.2.1\n\nmeta: version: 7.2.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Idcb0d9facbde2a516086a8d1195e5c2edb0caf4a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic6cb2d67dfe7026f809e31975622bde957d00b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2b41891370b11ff27ac83a576f33f4c137bb032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2b41891370b11ff27ac83a576f33f4c137bb032"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
