)]}'
{"id":"openstack%2Frequirements~779229","triplet_id":"openstack%2Frequirements~master~I404d599fd761bd970001aeff5760b95b3d96db76","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I404d599fd761bd970001aeff5760b95b3d96db76","subject":"update constraint for python-vitrageclient to new release 4.3.0","status":"MERGED","created":"2021-03-08 11:58:17.000000000","updated":"2021-03-09 00:48:06.000000000","submitted":"2021-03-09 00:46:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779229-new-release","meta_rev_id":"1a03e8bc8a1f4f2ca5d4e25c8f01179ba768ee84","_number":779229,"virtual_id_number":779229,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 00:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-08 17:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-08 12:04:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-08 17:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 12:04:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-08 12:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 14:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 17:34:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"24f528b754b92af6c01d881e52606db093c20aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-08 11:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c85d64ab1f33380b6621c7feb398c2ad74096f8b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 12:04:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"060f135edf143d5823a914dc45a61fcbec0638e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 12:38:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/370b3b7119c64c62845e2ad74c3eb999 : FAILURE in 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c34deba66a0729d55f493fe13f120860e9c7165e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 14:14:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91c63ef50d54e53a9c8dc7d72fce516 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f403aea964a42d18a89e91b5a764307 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e07e99f5f1164435bdf04d4ff1138de8 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd2e5c13cf749c89e77856af2067c18 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcc952189b084aab89353778fb7435ef : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bec390608f44a9880fef396eff71337 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7013e5ccc6374aefa89ba36125d7ec63 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d1560f29c09446559392efbcb803e108 : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/266210445e704747816a244981f2654f : SUCCESS in 10m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f14e608a79cf436a9baa5dbfe25b81b4 : SUCCESS in 14m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7be9e242ec234c09ad78dc7b5bdd3294 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f738158152a4162aa2c234dd56560c4 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4f1a1ff8cc944f9a71bfb44f1967a81 : SUCCESS in 10m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/039710872ac64bcc8b9ba816bee0855e : SUCCESS in 10m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d7bbfa30dea49858fa6e26a3390b95c : SUCCESS in 16m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ed35355989ea4ef4b0783e5db982e12f : SUCCESS in 5m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9c61c9dcde84214b21e63789e2c35c8 : SUCCESS in 32m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84172c817402487eaef8f2791a0f85a9 : SUCCESS in 12m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a24c10de37c2422e93cff0a1bff3313c : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d07525088e8244d59d0026416f0ee180 : SUCCESS in 21m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db09389c96f1476ab613b09efd2a9abc : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58699a5cc62c4a85ab4d2dadf3a6aed7 : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b9742002562b4177976e36c68454dc1d : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58847bbeed0c45df8faba8b972845985 : SUCCESS in 1h 00m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90e97bc106f74e38ae8f44dddde9ec62 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"21418378593a6c998ce8ac4a29d1cb582e7e59b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 17:34:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea9299053197825179174a8d90fd4e35cbd6907c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:40:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"018e31ba94140a3b0c051a5456261dd10f0fa5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:00:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc38f966a8124b508cd84c694a07c0df : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d643c9a34c1b4cc490fb919d4abcf9f4 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70990037f517498cad52dc96cb0696ef : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84f43630513342a0916075f73d544b88 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbab19d0c15b401c86dd8430a34005c6 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f84806590454f2db33c3d220b98dc03 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b77d5fcd96845c489092f7c5d264e01 : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4a11f8c55aa44769bd56549e92f0fe6 : SUCCESS in 12m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3478a80104624cb7ba13f5992b4bc9a3 : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/faa4bb397bfe4d9f8bb5768078a0fee3 : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09ff318b148346fd8a30ddb42e14e9fa : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1501295a61ca4402b824cec6318a9e0a : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07d3446c4160499296fa946a1994e1cf : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6d1926e291f54d02ab34bb26920bac4f : SUCCESS in 15m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9a25b8f6ace548359629c8e8cfc1e920 : SUCCESS in 4m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5854d6be3e234d779eed81269f3484b5 : TIMED_OUT in 44m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/52da16ad3b2446c68c75398fcfcaa0f3 : SUCCESS in 15m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3174db254c2f4660a86cf34050964a6e : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c390387fba943ccb3582066d5ce509e : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/233b7ecf22e444aea20e1095e85e3247 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa029abc4f6b46989507c46f115c9cef : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e4b9097b18c741649a78b3743eac9f80 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd68199759a1401380eec4f6436ac239 : SUCCESS in 41m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9620d9b774b4f40a685aab4e6ea4d58 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7e8621c67bff6c5ddadfcbafb6fb5732a7da4fe8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 21:26:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2828cef5f11734116ef8cf2b01dd81f2d77c5f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:13:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/809c3d5ea69143fa93c54eaaa954d0e4 : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":1},{"id":"53e14d51a9af61951fcaa3eccb331cebd3770b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f9f02d8ef144b39d2d895c0a210c4c : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895a97f8f18045a89cee1948406c226b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9dd980bdfca407bbbad62b62aa16359 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b82a1a1673475cbc16aec6a86dc765 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6eb5ebe367b4b388919a8cd38ea77e8 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad88519565114c029045ab26143c24b4 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed2b3c39f9d9417eae981bf25fe4465d : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52e8e11a68e148f8a570bfa70ca5d337 : SUCCESS in 11m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02006f97e3a54860a30b702e9a875a31 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/34db62f8a2b947f1999738228fc03c10 : SUCCESS in 11m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f7f54d8448414ab382fc18a6637c8e7c : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ce22e030a684d7db2e4ce594f09361c : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75519bb9f9914f358bbdd91edda3ac50 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f6b40249da741d79b3c9b5547395b5f : SUCCESS in 9m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5533ad0fa5f04e67bae2803b2c89f59c : SUCCESS in 12m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2dd4fcbe079842ea99ece57747207eba : SUCCESS in 5m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7145e0711464cf5b83e6e7e8ada2438 : SUCCESS in 31m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f604ba538e3f411ebd1d7660fb5a8602 : SUCCESS in 12m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b1bb2da65cb465a8102e0c429de9efd : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/91d62e0d876842a490b3f8a70f50d03a : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/90d635177ea44933a1763612e88866f0 : SUCCESS in 7m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2385fd97d659408d875a81c7e57011db : SUCCESS in 6m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6adbe8096f254bc293c28d2d20f9419f : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6ffc4c4b3b04e2faaa3688092acefce : SUCCESS in 45m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77ec93a438554045b83d733538955bd9 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"01cbcb51945b45c4263aecce348273d9285bd0ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df1b0b27d685a87a71832483db06f1ebceaf4869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 00:46:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd601d2cc1d64910adc8fa1fb8da010f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694f7471150e49dda849c96b8cff65cf : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6d14dd9235e43ce82fccdce60b3d9a1 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48536306c47e4fae872d74e561173be4 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/057a99e40aa046ec8571c08dd0e90c00 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e18d3bc7ac0e43f28c2959c94f1fa157 : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35dd5a25d01f45ac9ed3303e55599d15 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/691e5db54a92482bb60fba6d7993b8b0 : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bbb491d4c5e544ddbc7972ebbf9f2013 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44640229548841d69ec5b59dc9366ec5 : SUCCESS in 8m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0f117815c9094f3782d9aee9b54debf5 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de022df1e63b4201b4c8a8197f35b43f : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ea97ee6e70f4e1d89a2a1115a1ff5f2 : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/189cc2f903d5483bbb308046db3010cc : SUCCESS in 20m 15s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/763e49d812fd49e58e6a4e572f5c69b6 : SUCCESS in 5m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/06d7e7e02b6f4ac8bced133509cb98af : SUCCESS in 39m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3170bfe6d85b44e6bc42420d792782c3 : SUCCESS in 12m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b424b3fe64934d6491257bec4297f962 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5e89d60035149dc809248db2d2c7e6a : SUCCESS in 22m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65a062bb1cc3433c8c9c1d6da21fca28 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6feeae7efde24c16ae0a74d69b45bde2 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a19180aa585c43b3af548411cf01c180 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d34b0c9fc4bf4e6ea24200ea29617822 : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e88fa49ae384a29ae51415c56a4961c : SUCCESS in 1h 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1577f6bc444bd1ea674ffe6fbf3b588f11e1b265","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 00:46:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1a03e8bc8a1f4f2ca5d4e25c8f01179ba768ee84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 00:48:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d23242612145289d5c2e0890a1d82a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fe66ea13a4db75a7ea516fed1b516da12232b8b","revisions":{"3fe66ea13a4db75a7ea516fed1b516da12232b8b":{"kind":"REWORK","_number":1,"created":"2021-03-08 11:58:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/779229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/779229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/779229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/779229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/779229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/779229/1"}}},"commit":{"parents":[{"commit":"79ff19fa57044575ea93b57c29c0e0b46028989d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ff19fa57044575ea93b57c29c0e0b46028989d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 11:58:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 11:58:14.000000000","tz":0},"subject":"update constraint for python-vitrageclient to new release 4.3.0","message":"update constraint for python-vitrageclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ia6be75df1986537815f2438a7fac4ee4e522d799\nmeta: release:Code-Review+1: Eyal \u003ceyalb1@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I404d599fd761bd970001aeff5760b95b3d96db76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe66ea13a4db75a7ea516fed1b516da12232b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe66ea13a4db75a7ea516fed1b516da12232b8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
