)]}'
{"id":"openstack%2Frequirements~779264","triplet_id":"openstack%2Frequirements~master~I5422306697c0b56bbcad897a348e3173a9363b5d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5422306697c0b56bbcad897a348e3173a9363b5d","subject":"update constraint for python-blazarclient to new release 3.2.0","status":"MERGED","created":"2021-03-08 13:55:35.000000000","updated":"2021-03-09 02:22:50.000000000","submitted":"2021-03-09 02:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779264-new-release","meta_rev_id":"72c9bf0efbf105d9f1e9ef856b2f6212f93a3f10","_number":779264,"virtual_id_number":779264,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 02:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-08 17:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-08 14:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-08 17:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 14:00:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-08 14:10:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 15:42:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 17:35:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"003f1af709204f3270cc4380fc3279a88e960bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-08 13:55:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f793b53422e61dcd9982163576880e16f894ab97","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 14:00:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2750e0342588fba283c5a45d600dff00b284e49a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 14:10:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"92c77b84d4dfb4fca693c639b4fcf36f21976078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:42:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98059593a5014fe2b05fb2e5a9df569c : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad03dcc7fa34466a95339c9d18fba44 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e66ba1d9affc4dc2b80ff4c697f079c1 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffcd0c2ebc148848156f4a7df559068 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04c180aa55514bdda13cce83e717a485 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45f64857e7d4442384a14eff425dbe59 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b867a039be7e47a09f189280143de112 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87dc915939454241a8df20b07401a03c : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2cefefefe2314d46ae411f2b7937173f : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16200dbc86fd47a3ba5eed6749e74d8b : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/973cd9f25fe740109e92a6f1b64c5565 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9c4fe97ca7d4a73b73a07a8019e0701 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96e28dc98f0d41908119cece8ade0ed9 : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2fb88b5997654003ade9dc143e0397cc : SUCCESS in 10m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/beea90fee22b46ff98131c3d57be2d5e : SUCCESS in 24m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2dc1f12691484859ba46c1e865dc1536 : SUCCESS in 4m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/07b7864790d34b42844204f48469b234 : SUCCESS in 32m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce00ec1a28e34ad2832b77aed35cbc71 : SUCCESS in 13m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6c171a3aec54958a7aff4c3f7bd7e44 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c607dd76907e4fb9afd5ba0355ea011d : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77f73817bc8a4c898b3a8818ef8fa65e : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57b60691ba6e43e891493ddd4950ce7b : SUCCESS in 8m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/823eacb22ce549b4b6866527b1066445 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2702b1ddb09445f98a1ab54f1413d3fc : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36cac7c45bc34ef3b6d9f8ecf1e4e86b : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b30a5c60f867b945fa8b6faed09f96a9f3d3a701","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 17:35:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77e658bb3e9e833edfbeb2631f0f93b1897ad917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:41:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"354933c36fcf3e34745386420629bae1944137a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:18:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4ecb5e6200422db70b47dcd9ce28ca : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9270cb9cd6b649479dd680957691f860 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6984c5e87bf845c99fc32d184ed69971 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f0913ede3aa4da294bb2672add8c5fb : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4a47be1e7fe4828b71fb71aae1fc4e1 : SUCCESS in 8m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a62e8b6f15f74a9fafee05e9aa6216a4 : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b08c8c3bf9084d53b00fab0f7ff8793a : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f6a11669c8741da93ebfe782d27763c : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f5903544a7d34003a9642558d8711eec : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eff7af9e453d4466b67c837666d59563 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/001058b0c59745a2a97e845e0e7be0f9 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48a8d5285caf4499a48619e884890c66 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/66dfdfe15bfd4af7955c9d1a5ccd10b4 : SUCCESS in 9m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/32aedba1c855427d85a3cf8781b7dbb9 : SUCCESS in 13m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b97b58a9bd304b5fac35beb138b9a151 : SUCCESS in 7m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8056f33c2f0a4fd0a8f150484ed071f8 : TIMED_OUT in 41m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49ee4677253d443f8ded7d459b2e416e : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a5535f1c4ec4992bded9266d4dc6875 : SUCCESS in 6m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e1220071bd843d095e4740b12647da3 : SUCCESS in 16m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/22de59baa6f143aeb87f228f5bf8faf5 : SUCCESS in 8m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e0795b19ef94f6685efa76b9448adc0 : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1f0b0a013b54486d8292ceabc5a3fa1d : SUCCESS in 16m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb9246eba0554a929bca08e24bada19c : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8969d0446e23460298de82e2f9f2a203 : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f78c9982dcd41fb36be917cd004a3873f535856d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 23:15:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f4c2878adc4cc087e3079dc47aaee81c28ccfd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:44:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/661f1a709d2d4530ab38f84ea3fbcb5c : FAILURE in 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"022a05d1035be68082e2384cb321ac298512972d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 00:36:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f917e07f9f794c21878fe82611072d51 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4c64e6a56041d3aec68009f9c79c11 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17ac1028a4f24d15b25ff15fcd5e6b83 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da58aa03368d41c787c8a0f8106c53d4 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e131480525fe4646b9bfd480c1f30381 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee0aa02fd6e744ed9bb7b154ef902b11 : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f59d2e67dcfc4b45b5ba9992764ded5e : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e4604b553994477b6e82e27802d97d9 : SUCCESS in 7m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a0ca25c46e6a4784b6818a9cdaa7dc8e : SUCCESS in 12m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ab3694fff227424b8396d1601c654319 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9e8bf249038a4433b9a57a3701ae34ae : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/443d870a2ef04e799114a500c6aeff17 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/662aefb24b064adaa940271b4d5f4d01 : SUCCESS in 7m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22507bc5298f4d968f9b321828f18436 : SUCCESS in 13m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/73c164a351dc4e8eb2b73530567b0419 : SUCCESS in 14m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/330535e5067d418186320396ca61acf9 : SUCCESS in 4m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c80731a83bf4facad3038227ad1c357 : SUCCESS in 34m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/abe0a7e2b17f49b0ad11c37fefc8f18b : SUCCESS in 11m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5d2637127f8486b8d4d0e50abbaa214 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0a6e5dc6d1c40e982b8d128e3264d49 : SUCCESS in 16m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51acd3616049490f9253ca1ef89ee48b : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5220d3c2f7d84505bf14512a6c6540af : SUCCESS in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d6f561790b34453c913378ac454cba5c : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5e0bb82dd5a4b189b0421dd06f3a3e9 : SUCCESS in 51m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01964f7a864e4d6db43034cf16e7b967 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f3f5bb4c5cbd43b76725e31edfe098272e48affd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 00:37:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"610c4e1d6005c3802895be6f3860dd0309ebd83b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:20:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab372b565864b17843c2f89874fe976 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256bfa5086184f4cae25fdf4c38d1705 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fba97adc8274417a1720ad1a14ad4ae : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ccc618dd49245eb8d3499bb30df8740 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62f8b26d24444989baf556595f14950b : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/685f679284aa4dc9826d3a438b0c8d35 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b9e50691bbc447f888a39c1977ee1c1 : SUCCESS in 12m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2b04ae95894e4787907184bac5f72738 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dda77d1b398c4421b5a05f4b78d426de : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2bfdee5fed1148bc848a85dbcf0074bd : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9926e24e9cb4e0c849e373b7efa49fc : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/588e07a1e648420ea6027b3b665c5995 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/237b37af3c9142a99f57880fcc64adf7 : SUCCESS in 10m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d76aad4b1f94affa69e16f9694cc629 : SUCCESS in 13m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8c41a330e5554648873c7f6eeb066ee0 : SUCCESS in 5m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02c43f7e6f0e4d548c6a06482515db1f : SUCCESS in 28m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/77143f9ec4e640d0931f6aaae46f294f : SUCCESS in 12m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13dc47d30fb1455fb0a48c8afaf2b562 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37926a6bd4654c089fb7af924840d4b9 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9756dc1d3a84871a2d93ed3041a5ba0 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4a876fdf603405a98412e30e50a0bad : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1ff3e07b7af44f318dd0053760fac450 : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61c5bf4596a947bdabe3fb3034426d14 : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcfb7b7b748a4f3288257939cae427b7 : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"180db12a9bca670bdcc84f6955512c34ce874a1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:20:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72c9bf0efbf105d9f1e9ef856b2f6212f93a3f10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e426aa46f64958bae37d6936dfe79f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23de02c07e8fa7f9f66f06b72b39746398b63e13","revisions":{"23de02c07e8fa7f9f66f06b72b39746398b63e13":{"kind":"REWORK","_number":1,"created":"2021-03-08 13:55:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/779264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/779264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/779264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/779264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/779264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/779264/1"}}},"commit":{"parents":[{"commit":"79ff19fa57044575ea93b57c29c0e0b46028989d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ff19fa57044575ea93b57c29c0e0b46028989d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 13:55:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 13:55:31.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 3.2.0","message":"update constraint for python-blazarclient to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I490a935c6220217da5791e6dd4566303e696d9cb\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5422306697c0b56bbcad897a348e3173a9363b5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23de02c07e8fa7f9f66f06b72b39746398b63e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23de02c07e8fa7f9f66f06b72b39746398b63e13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
