)]}'
{"id":"openstack%2Frequirements~779430","triplet_id":"openstack%2Frequirements~master~Ib72800f13e33a21cdacb99646ba50577514663fd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ib72800f13e33a21cdacb99646ba50577514663fd","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-03-09 06:35:45.000000000","updated":"2021-03-09 21:55:57.000000000","submitted":"2021-03-09 21:50:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779430-openstack/requirements/constraints/noclob","meta_rev_id":"c96b79642b5afd57417e3647ccea5eb184d3032a","_number":779430,"virtual_id_number":779430,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 21:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-09 16:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-09 16:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 06:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 08:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 16:25:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0911637e457927a59259df6b13d7759ef69b1de5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-09 06:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"384b3438f1640bc7ebe80168414921a9fba2aacf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 06:47:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5daf7977f3be4db09b608eac21fd5412 : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4a0b810fd938637d6de0dcfe702d912016c2fc53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:45:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c4a8f943744602bc408880a9e243c2 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a6e324eacb4801bcf54c2d7dde8e48 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c45ea7f6cf134c52a6e2a991eb09dea6 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a0042e1aa464987aa35bc44eeb6dd18 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38aada3ba7bf4bf1969af69458f8a2de : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db1e18b6044d44e4a83979a91920319b : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b2a0a7c3b144764a958d55ba1412d55 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f14044baa2894299abe77fd173e26cb7 : FAILURE in 5m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c8a791d061a48949407627e69c83be5 : FAILURE in 5m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff336ea4b9d14a5aa8fd0de1db5b5967 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad408b9babd1499cb17a5d222cc6f730 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1524e195705d44ee8f91064c8847c9bb : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/647c5c3501cb4be6bfeeed66719c1708 : SUCCESS in 10m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07c007dafca148b8816063c20d9edd45 : SUCCESS in 10m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5cc417101bc54fd6903e5d77269b7bbd : SUCCESS in 12m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/4dadbd46114945cf9a85e72eb7414de3 : SUCCESS in 4m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6945a74a9fe4b0a81ea2345869c6fa5 : SUCCESS in 29m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/520d66cf2e29496ba2c0d555fdc45806 : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20c7b3373a834ae5958cf1f8448426ae : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c58e4d825e7e4bc099d4229be433cd26 : SUCCESS in 16m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a933bfbbac50462289f6d93b6873be0e : FAILURE in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/387a129fec654017bc5e60b962bb62d0 : SUCCESS in 7m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/da99e97cb4c942f39674b614f27c9d57 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6c523f80ab3492e826f5bd777de0a6e : FAILURE in 1h 11m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e91562e18ce1423aa8a8846b37dc12b2 : TIMED_OUT in 2h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9670ef4516c92387231ea149b1faf8bfab21dd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-09 16:25:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d99b3e12aa16b73f1a656d11d216ce5166e6c58f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-09 16:25:52.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"454dc7aebef899f0703d49adaf8a22a06ef25519","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-09 16:26:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3204bee583a6a295a0b5ff403ed0752a76ba87e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:02:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2978d2a5fdaf452fbff58bcb556f0157 : FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e984ca95f189ec09eaa275592b5499e081607974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:07:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4137eaebd7874ff3ae2255f9d214c2a2 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c317ede65a7342f59e46136489df157a : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fd626d6c68d4720a83337b9b6c09dc8 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff46210b36a44fabfd76f689b2a6297 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26bcd0d651f6495597e8980a0d04a480 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0ec47e9611f44a1a3637cb90f7fd71b : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3aa21f31c70f43d680236b17db903e7f : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9378675e017d48c1a28f23fdff3f2dd5 : SUCCESS in 12m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f6bc960eb80406a80bdf0b220c3e571 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/43a9de67957149e9a79ced1d5702e444 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7a0e767361454bd497dc3c73bea962ea : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/492dbde762ad4711b095894a5bc7a5a2 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a4920055c004551babe8941370b4c40 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98ab49ce152646aabf994c2546d03659 : SUCCESS in 10m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7476ebbf191d41cba3fc20882e6c204e : SUCCESS in 16m 03s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/da7ba4209e754a4d830c041d5ce499e5 : SUCCESS in 7m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e3601f7208484554a1759f69bba4d17e : SUCCESS in 40m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d2a91f7a61a0422788803d433952be27 : SUCCESS in 11m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd93e32a062346aa83396976259adfb0 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ff4cb1c82894d9b962668698cbbd5ca : SUCCESS in 22m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3a7832bda51d491f9a7bf33ccc3458b9 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83de920a96604586b892d376e40c978a : SUCCESS in 7m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e502cba4979148d1916802b4bf9b2d85 : SUCCESS in 17m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff96b0f4e14940cbb36a318615488d12 : SUCCESS in 1h 15m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b07fcec7f2f43c985be00695166f78f : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a4b1ffb79a815fc37df5074d112b47977083c085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:03:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26eb6399850e89450452c718e0f5fc263cd636c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:50:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b85871f0b394afa93797c289d59b471 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec76b463b064e17aae5bea0c920662f : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0fe61ecaeeb44068a265fafe398f4ee : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c2d3c0deb0e48c7a90624dc64eeb230 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98dd6c5bee964f1eb69c6dcea2a9c956 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff8ea149235d4a1bb714e54d21b26470 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d1d41348b9a94a7bae32d5bd8a9b7cdc : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dcbbdc27ad1949fb8f9d00376eda0a8a : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc10952b444b422783648fe4925c293a : SUCCESS in 12m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62824a133db2435189452d7d5126a91b : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3896b58385b84f5194d623fb09c3e5b6 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7d1edca20a3432d819e47b1aa199e02 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7e0120be67c4f8d8ccb01d922a03738 : SUCCESS in 10m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c1057ff7a3f7434c8cc2cc211a87bb7c : SUCCESS in 14m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9a1d9d0c327843a6b05f7de7b7891a64 : SUCCESS in 4m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8f0b104fd44f4bad882391357dc7f6f3 : SUCCESS in 31m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3571da4e11194b108988fd989b328477 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/087fec6a41a64016ab6f467315917bc3 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adee181cbd9d44a198dbf3518fab13a0 : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0bf374c102084086948442fdf95665e4 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5798a31ddc14f3b96bf70399e249f85 : SUCCESS in 7m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6bc7904b8ef84f1f8b5c4aaeb0c85c68 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccc8d233a78446d3834e70d090da82e9 : SUCCESS in 39m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca7bf2d5eaba4d3bbb8fff0c2a6b22b3 : SUCCESS in 1h 40m 10s","accounts_in_message":[],"_revision_number":2},{"id":"311a34448b17f736e1178f7fc74dfcfaa460e241","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:50:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c96b79642b5afd57417e3647ccea5eb184d3032a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b213549786416eb06201a87ab1c4c7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb252d7ee78abd31c3ee3dd5e9ec724ccc3919f7","revisions":{"801dec176afe20d7f5162ece9242b1bebc13a66e":{"kind":"REWORK","_number":1,"created":"2021-03-09 06:35:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/779430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/779430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/779430/1"}}},"commit":{"parents":[{"commit":"0395cd80f07180ec5549df5575fd6f7233097eac","subject":"Merge \"update constraint for python-magnumclient to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0395cd80f07180ec5549df5575fd6f7233097eac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-09 06:35:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-09 06:35:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib72800f13e33a21cdacb99646ba50577514663fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/801dec176afe20d7f5162ece9242b1bebc13a66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/801dec176afe20d7f5162ece9242b1bebc13a66e"}]},"branch":"refs/heads/master"},"bb252d7ee78abd31c3ee3dd5e9ec724ccc3919f7":{"kind":"REWORK","_number":2,"created":"2021-03-09 16:25:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/779430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/779430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/779430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/779430/2"}}},"commit":{"parents":[{"commit":"0395cd80f07180ec5549df5575fd6f7233097eac","subject":"Merge \"update constraint for python-magnumclient to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0395cd80f07180ec5549df5575fd6f7233097eac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-09 06:35:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-09 16:25:04.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib72800f13e33a21cdacb99646ba50577514663fd\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb252d7ee78abd31c3ee3dd5e9ec724ccc3919f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb252d7ee78abd31c3ee3dd5e9ec724ccc3919f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
