)]}'
{"id":"openstack%2Frequirements~779437","triplet_id":"openstack%2Frequirements~master~I2d473f5a2d41c116faf1b3c700552680610a3999","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2d473f5a2d41c116faf1b3c700552680610a3999","subject":"update constraint for ironic-lib to new release 4.6.1","status":"MERGED","created":"2021-03-09 08:14:48.000000000","updated":"2021-03-09 20:34:29.000000000","submitted":"2021-03-09 20:24:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779437-new-release","meta_rev_id":"3fd80bba8b3cb4f7bd35dde86df14d3b3117dcc8","_number":779437,"virtual_id_number":779437,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 20:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-09 16:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-09 13:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-09 16:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 08:58:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 09:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 13:38:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-09 16:23:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"130c6f2566779a7e2994d0a3a309b139c7ac2abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-09 08:14:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d03ff6ccc73724c0b78d0bcb5d85248014caffd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:58:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7662ec4789cc4247bdc796dde8867a07 : FAILURE in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"12954e67e2b27d263177271c899d98936bb9eca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 09:45:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161e2c368dc746bab83b3dff5aa609ef : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f8bf26d0a5418bbcfe6541a1f89843 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994f788a9f714d58bd70390e5f12824a : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/826742a655e143bfb96aae37b83ceb8b : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ece1512d8b054a19a897ca6062901c9d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3285e0e6ec94159aaa39227dfc84cda : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52152418c5554f5ca67c0558ab5d0806 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ea701fbb1fa442a8fca4d53bc0884c4 : SUCCESS in 12m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c400893688b64bfca3cac4bfca17c96c : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/57ff8fe1f8564aa69701d231bddc78ec : SUCCESS in 12m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f009ff5221204aa5b24b9585ab4b7d20 : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e63519e078764361899a2d94a9f1f76c : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d5df22ccef54631a406cf7536177333 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/da4682d8bfdf48eba0998e028f8a56e7 : SUCCESS in 12m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9132fca85f4a4c46b0072ff94b7907ca : SUCCESS in 15m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e44188e9c45b4c319157979c492a6e1c : SUCCESS in 7m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ebe186e7537c4740b8d031c1910ef69b : SUCCESS in 34m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79cd1ee2b4c5411b9b141824682aec49 : SUCCESS in 12m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f56d23cd69f4591b3964fd96c2721cb : SUCCESS in 6m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f369bf606154939923dfd6eac68bb13 : SUCCESS in 16m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02cf4722841646e9b4c6f7e44f06feb7 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/031ff664445a452e90835c24ce62ad78 : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb0838a573324886814c492d462fc7d1 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dda0469beef94799aff2878019346dc1 : SUCCESS in 59m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0448f7bc6864549a9edfe7d9fabc4ee : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e472503d9f22f6e3aed6ec44148cf6c64e60bf3e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 13:38:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2be19db96d70900a9ae0d21c3de025f72bdcb237","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-09 16:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2988020923c6cb23bfb3f326d575c2ced766b60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 16:24:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19150e493e1a7629ae75c07e7f88c2fbd287a865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:24:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a3d4ae2c3d457887cdc5e5853cf9b3 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd2ab8945d34e06827a3fa27f2aac12 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ff57969e7f49f997f61f758dcd2057 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae3334fe130f4f8abfe13c6c4574fd74 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15cfe008e6b648fda04527849cb1cfdc : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d2993a1341e4008b14088997d61e81f : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/594dd766e815470ba1e89ce1383ff742 : SUCCESS in 10m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/708d23b46c8b4e2da08e169c3f1523d6 : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a7a5035e9c24b18b7764c6b12edf889 : SUCCESS in 10m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a9d2669521fe41f685da9ee847ac9a98 : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ded04661f66d4498a8ccf1b7473dd52c : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/830df0c3b51145f3bcec7feb74181308 : SUCCESS in 9m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5934234e08f249e187e5611cba0c9471 : SUCCESS in 12m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4ec379175fd4822b5f147e249023efa : SUCCESS in 18m 27s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/72da08db46f641d1abfaf5b6817ec447 : SUCCESS in 6m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6e56824caa046bc833a648147b4b004 : SUCCESS in 34m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13af4b4af44f48efaae0c400585453ef : SUCCESS in 14m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d865b71344c412a87f11b7d28fa7aaf : SUCCESS in 6m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c4ac6742d0a483abd229f56c5910d4f : SUCCESS in 20m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cebe380e9cc343beb1cc7f16dde3c526 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70de2cd4e9f24564b35c18a167baf125 : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09885d73af09488bb56af288d9a9ac5b : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/755262cd350548a3b26d32fa675b8b8e : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f77833fb4c88452fab352985a6bc3d3e : SUCCESS in 1h 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5cc4147688f8bdc17888db89587702112f8f8947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:24:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3fd80bba8b3cb4f7bd35dde86df14d3b3117dcc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:34:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94502b42a6d4086b294c45dae9263fe : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32a0ede07f41bc203bf06f377c1935f8d266e320","revisions":{"32a0ede07f41bc203bf06f377c1935f8d266e320":{"kind":"REWORK","_number":1,"created":"2021-03-09 08:14:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/779437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/779437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/779437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/779437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/779437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/779437/1"}}},"commit":{"parents":[{"commit":"0395cd80f07180ec5549df5575fd6f7233097eac","subject":"Merge \"update constraint for python-magnumclient to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0395cd80f07180ec5549df5575fd6f7233097eac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-09 08:14:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-09 08:14:46.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.6.1","message":"update constraint for ironic-lib to new release 4.6.1\n\nmeta: version: 4.6.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: Id23734d7078c61e4278ee930e67efa86fb6a0a77\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2d473f5a2d41c116faf1b3c700552680610a3999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32a0ede07f41bc203bf06f377c1935f8d266e320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32a0ede07f41bc203bf06f377c1935f8d266e320"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
