)]}'
{"id":"openstack%2Frequirements~779701","triplet_id":"openstack%2Frequirements~master~I657d6f84a50252f0be45d4ecd01b1a397167f259","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I657d6f84a50252f0be45d4ecd01b1a397167f259","subject":"update constraint for tempest to new release 26.1.0","status":"MERGED","created":"2021-03-10 09:45:30.000000000","updated":"2021-03-10 21:21:21.000000000","submitted":"2021-03-10 21:19:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779701-new-release","meta_rev_id":"878fc5051ab3b3005d424e00575fc3cd5d592aa6","_number":779701,"virtual_id_number":779701,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 21:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-10 16:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-03-10 14:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-10 14:21:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-10 16:17:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 10:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 12:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 14:21:38.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-03-10 14:24:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-10 16:17:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"541f28bb74429d040ed48f2c67af553697c1b4ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-10 09:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edd09707a845500618323352329c4a3a8b20f1cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:20:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/44643e0e86384a3fb0c5118ca24143ff : FAILURE in 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1058810b544d07f941cbe7795570454bb10f9a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8e04f4fd6a4812af3c3f8221c1e904 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5bd545951b49c48d849e9417b05b38 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68e468b6feb43b280b9eeb6a14ed457 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78bb02f7303c4134ad41c560e63f1f77 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc584689432c4614b262760d0822d210 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b41783fb65c446fb99461e26e84e9e8 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b81a64ac33b4ad19b201d424723d49c : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64d112cf682a4954bd364881bd8fc5b9 : SUCCESS in 15m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a39afa531f842a2a4ed163dd31afdb8 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/884ecc72b5c048d98c539daf103710d4 : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb54dd64eb0543cc980a2c78b2666274 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/204cc4db201545bdad7392d46d92edca : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a68383ba8c5a4a5caf7adcde05f8086b : SUCCESS in 7m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/806cfbd68900479cb0462447c572c344 : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f2d3a5739ff42e19e5714dcb6815fa9 : SUCCESS in 12m 47s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9321ae202e67493fb8d8436d8ae92d97 : SUCCESS in 5m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fe787ea9cb94a1097365c784aa8050e : SUCCESS in 37m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4df267a415af4dada357b5c581d0d1ed : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69fac0ce277b41f584cf8edc62d09e77 : SUCCESS in 6m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77489809406e488e8bdf4a41b00f0f56 : SUCCESS in 15m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f8169c14264043a9b648456ac08f73dd : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7adbb69fe52e4a1a89baf3802743a150 : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6100bb7e00ae4b4a83c0bcf3f603ce4d : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60e620bc8a8347c1a6a52dc6105c520d : SUCCESS in 1h 11m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5da05e96d5044da8d8bb80c98277b00 : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"699e8b93f65c54f813bebc4d2f03a20b2a69cd2f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-10 14:21:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f1fe178779f50cdbddcb52b97fb44791a8ff5af","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-10 14:24:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e0a62edc6c5ff36339fbee2967f8235fa318b767","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-10 16:17:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ba618e9b38166cec2044f6e51eafb19c1145c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:17:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7a2ba3d1b782a6e34e1ee2c400e7f6bf1db99bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 21:19:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523daadac852420e9ddc714e5fc338ca : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef2e717d4194c2f8179ea5b1182bd53 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/671ae610193b44a59fd20b610ec33fb1 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6acccc4ecca4aae988b43192a720a9e : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0da8d2dc6fc4bbca45b808e24257f6c : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfadf62524ee46c7940c34602ee370e5 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c32dc5fab6a1447099675e0c05fdc185 : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/de1a51063370457492ca44f98986528b : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/08eed7c83273480c80cfdc209c8c1816 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49c767887b024f3183c1dd5d20532abb : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/701bd052c33e44a4b225822da6dc98f9 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5acce46cbcc147d493d3480c7d4290ae : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aa87542caaff4a5ebf2a4560408d1efe : SUCCESS in 11m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/882162ecb7194f59b1ac21f336bc9241 : SUCCESS in 13m 35s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/25a662e5c6f94e3c8f6b671646ee5f12 : SUCCESS in 5m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/27be5a20fb734665b78c3b99f92041ef : SUCCESS in 30m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/004f2f8c179f49c6af778f35a8c76259 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4078d05771da4577a9c2ae1df668acf9 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e4dbfb76f4c4ad1aba5d1d8e1af7c80 : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b1df8b121384f398a83a5f4313b30cd : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbbcf5272c1246fdad7eb2b22519bffd : SUCCESS in 7m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bdce50fc6314fd68e69f1d32d133307 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bc39647c0bc43289aeec104bbdbc710 : SUCCESS in 56m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e0db8cb92994797b09022f96600f5f3 : SUCCESS in 1h 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bcc8e2bc82fb8e25d1a7d50e4729945bb01a5c30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 21:19:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"878fc5051ab3b3005d424e00575fc3cd5d592aa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 21:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b68ef7aec3b4db1b9d5fa43f585cb26 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0c30d1a12b87a4424ed4842486226b08cf7d3f0","revisions":{"b0c30d1a12b87a4424ed4842486226b08cf7d3f0":{"kind":"REWORK","_number":1,"created":"2021-03-10 09:45:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/779701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/779701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/779701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/779701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/779701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/779701/1"}}},"commit":{"parents":[{"commit":"202a7efe0bc4ba8a78a4196761a8fa2c2224f085","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/202a7efe0bc4ba8a78a4196761a8fa2c2224f085"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-10 09:45:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-10 09:45:28.000000000","tz":0},"subject":"update constraint for tempest to new release 26.1.0","message":"update constraint for tempest to new release 26.1.0\n\nmeta: version: 26.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I691ceec1037712b3e01d52777c2dbb1432efde20\nmeta: release:Code-Review+1: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Masayuki Igawa \u003cmasayuki@igawa.io\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I657d6f84a50252f0be45d4ecd01b1a397167f259\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c30d1a12b87a4424ed4842486226b08cf7d3f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c30d1a12b87a4424ed4842486226b08cf7d3f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
