)]}'
{"id":"openstack%2Frequirements~779805","triplet_id":"openstack%2Frequirements~master~I0a3f8c354f41fa0c19fb9d60ca2b63d8b8bfd744","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0a3f8c354f41fa0c19fb9d60ca2b63d8b8bfd744","subject":"update constraint for python-swiftclient to new release 3.11.1","status":"MERGED","created":"2021-03-10 15:14:40.000000000","updated":"2021-03-11 06:22:49.000000000","submitted":"2021-03-11 06:20:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779805-new-release","meta_rev_id":"2336a976085e6262931fe10d440d5ba6f5a810c2","_number":779805,"virtual_id_number":779805,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 06:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 01:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-10 15:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 01:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 15:17:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-10 16:04:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 22:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 01:46:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"75728d0eeb6898442c2792469ee9c4ac7dc6092c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-10 15:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fa76719d7f0ba1378c2ed9737d8d89865928456","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-10 15:17:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3deb2930b5e41a7fc16e7d58c1bbb96caea4fd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:04:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d77c94289e34810b6ebaeb1512c1cf2 : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bb103aa76504b809d8aa5461d9aadf9928e4ef92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 22:10:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e558b4e6930e4f2682d9a37edee7a23e : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2c8b8d4db44729b07240a72f7f6272 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d76c6ddbb564f54bc47377afc40d2cb : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/409ef04068ce4e0fb8ee8dfae1a92034 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9ef5b1a34e0452fb242c85abec23cd7 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32d0b638f86641d39d4d208435539720 : SUCCESS in 7m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a188bfbd42de47d7a2af9b45ef8abe4b : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d126cbf3f6f45fcaf17f73a4e054e0a : SUCCESS in 9m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c18f555fa7e94614b22c3cae6f38f734 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca7c7487f0244293860f4a30a3235006 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13764b11c37c4d5aa33f1c4faef98106 : SUCCESS in 7m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6892347acf314eb58592f3de2345fde5 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bea076220654851bd7d2dac28398582 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c2e3bd18b3749eea0bd647b9dfdc8f1 : SUCCESS in 10m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c72421cfd284878b2a98b4180296640 : SUCCESS in 13m 34s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/986f5b5bcdd14568ab50b7960393df3a : SUCCESS in 6m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b3cc6a9dfe634d68b695b7a3418fd25b : SUCCESS in 37m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5f7c619b7cf4a55a67e6e1018d468c9 : SUCCESS in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5dad6604f1d9499681d477853f06bf07 : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8adbb3026104406aa8e06e2110f6d7f8 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/600c9dbc1e9c4e42b7b9f9ea03786ad1 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a790b59d7ccd4c2f8c4a9fb3c6a89ede : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4c4f79bd8ce74c8c899457447b054eaa : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14cabddfb85647128f439efca50e387c : SUCCESS in 51m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b04236f47504c90924aa006f7401c9f : SUCCESS in 1h 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"19dd087ad396b6e8f45feb9a6f2db767df29ee7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 01:46:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3010f82f0c83b89fef32f7e4fab3cabc27504b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:47:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee6897c699fc90923670db3c5d9ee0b117485cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:20:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72eb00ec20b143bcbcffb3b0ec3f7814 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d773d838c3bb41f98fa1afb1fd0a3bb9 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f91414bcda148979ee9d379f20ac3ec : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b34bbf6990c484581c471a8e58760df : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6561af9a4e1c466eb85d04c0fe4ba8ae : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c817f06c0a249589b8a7962fc576610 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4d8ea6b11ed48aea7d5823131b6c84a : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/425b1ef781194a17acd2bf025d4008d0 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc4bcdaaf06241679ce1f8e109354f23 : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6727ca1e7b244f68ac6dfe03237c573b : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a7b8727ee27a4bec8beb887ccac6c694 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4c425b0a8394ea888394ae44a08ff58 : SUCCESS in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82a6b3d1a8f9430483600568fa6b9ca4 : SUCCESS in 12m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/856407f9b28349a69fef6477ffceed21 : SUCCESS in 14m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2e89c8d2339d4dd6a206185cd779337b : SUCCESS in 4m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18aedb2c6ff843518683e403ef3e915e : SUCCESS in 32m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b593ac8fa95451d854fa4c094c00cc4 : SUCCESS in 12m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37710d51a98a4823b0cc744cf7e24721 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ad3962f76124ad2b8a7d313c0a2d1b0 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d50a31e645234692a15fc96fee49ea57 : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3511e5ece00743c8a62eb70369f879fc : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73bcce8912be4a09b58b67fa85256ecf : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c14bf09c71a4b81be8b3fb87026f663 : SUCCESS in 54m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fd720f5d899410e8866a7626b59a7fb : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"075b72b3cd1b262c4802c8b3157e1ef1d3b3a65f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:20:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2336a976085e6262931fe10d440d5ba6f5a810c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7900c956d7b4723923ccb8982471a2f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9738a8d0a712d49234e490f099a8a28bd3cd0970","revisions":{"9738a8d0a712d49234e490f099a8a28bd3cd0970":{"kind":"REWORK","_number":1,"created":"2021-03-10 15:14:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/779805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/779805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/779805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/779805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/779805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/779805/1"}}},"commit":{"parents":[{"commit":"202a7efe0bc4ba8a78a4196761a8fa2c2224f085","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/202a7efe0bc4ba8a78a4196761a8fa2c2224f085"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-10 15:14:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-10 15:14:38.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 3.11.1","message":"update constraint for python-swiftclient to new release 3.11.1\n\nmeta: version: 3.11.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Tim Burke \u003ctburke@nvidia.com\u003e\nmeta: release:Change-Id: I0eb722fe343f191a60630a68bb7048a90dbf3fef\nmeta: release:Code-Review+1: Tim Burke \u003ctburke@nvidia.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I0a3f8c354f41fa0c19fb9d60ca2b63d8b8bfd744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9738a8d0a712d49234e490f099a8a28bd3cd0970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9738a8d0a712d49234e490f099a8a28bd3cd0970"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
