)]}'
{"id":"openstack%2Frequirements~779907","triplet_id":"openstack%2Frequirements~master~Icb07708142f54e89178ca4027b5afabf4dea61a2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Icb07708142f54e89178ca4027b5afabf4dea61a2","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-03-11 06:18:09.000000000","updated":"2021-03-12 19:21:31.000000000","submitted":"2021-03-12 19:14:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779907-openstack/requirements/constraints","meta_rev_id":"79f55caf0c99b9b948c60f69f246329914f841f1","_number":779907,"virtual_id_number":779907,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 19:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 06:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 16:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2021-03-12 16:00:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 06:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 08:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 06:27:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-12 08:55:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2958aea33f0c83b296c285a8d43e61152ac22981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 06:18:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b1f564d1c4559cb551a0fbe60d4fc77e5e2a05d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:33:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3785be3e243f465d91ca4455d408c353 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e5253efe10cfcdb54b9a374faecd0f8b5316174d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 08:36:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4010709577452f9edcc847d3e60941 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbdedd5473184ed2981977012bd8d4f2 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee58e678f3047798643d4a5c3906a32 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/244645d3b7aa4bd99d662f6977298d2b : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/743346ee91b246759ac3f3ea51f40f67 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c53d42dce5649aa82f2f6f7bb2681ad : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cbe227f1c184bafbec3715f96070bb8 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef358f9843d44339b3cf56675fac1532 : FAILURE in 7m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77b43431f4264377847dc138887e11eb : FAILURE in 8m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7d365d30fd040c9a401510b4e20fb10 : SUCCESS in 12m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b307667fb3074fda8eaf52c74c6c9a91 : SUCCESS in 7m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/595d906c44d2418691060253561c0d34 : SUCCESS in 13m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c9a0aa43d1c424790b51e39fd75e83f : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e477ceac1cf34982adeb067e137b9cab : SUCCESS in 13m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a86110ca998a423784ef34786c634592 : SUCCESS in 14m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/66022669656d4eb4bea8aa8f2dbf45d0 : SUCCESS in 6m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6baea307dafa40ee91badfbc4cf1d5aa : SUCCESS in 38m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7bd268b5df9d44849b17cb785c9b052e : SUCCESS in 15m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cdb5b34520d04d98b9210d47197aafa2 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d873add57a7d4e3ab34868d4ac3de45e : SUCCESS in 23m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/95275dd9b5ac4f14aea78ec0d86180ed : FAILURE in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8916041c6496437399a6474d7ae9d667 : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9848d2cfa112458da1d4a81cdc95dadf : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a98d16182cc948fa8d4d11b64fc280d0 : FAILURE in 51m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fed4a40ca66478d9d13053c22a4e4cf : TIMED_OUT in 2h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"50bf2ef80d88cc6eb52c2be6349147f86b0eabbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 06:22:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b75b4b5068030ad4ea08411061f5f2da711f3dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 06:27:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bde1246a515ea6307cdc150d668ba2ceb63f30a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 06:27:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05df789f036e73399103cc259c55649df62ad61e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 06:32:06.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"28a1f1fdac8b45c570ca8a1132a29d084a63c5d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 07:52:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9ff437c98e4b20844518bac7429cb4 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f37e7a0c494446f82e61ced7b4dac3f : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684b3dc45a0e46d8bc23b24727634d72 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4af6ef838e394edebb75f60fb36dbe27 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19f85ebebff944d6ad6a7f3e46e3fee7 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f59f414aa744b1f8ddddb69a99de0e1 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c325bcdca03048a790719102de626fb4 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af5fac38c46349f0a94f2e62c97b2630 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/154f0b834b9b41719ad2b6388b8d9133 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8738c74b0bd647189dc1f89501e12b3c : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c9c90f1c307f424c9ee0cf6df55cddd3 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9afc57e81b1a4657abb9572dd20dee6f : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71c534b2d7004a2aa31a55212a5b269c : SUCCESS in 6m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4bee141dac0445e6a7feab03907c7fc6 : SUCCESS in 11m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cfd5149f23f44c938aa197d180fd7bc8 : SUCCESS in 17m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/791c025e6699409390c1fd2a599651b0 : SUCCESS in 6m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6271604d37944d4c951b1be434f4f670 : SUCCESS in 32m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b07b983210914b02822b646f96dd9951 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6c6f6ee624b4fab912e789f5e5da0e5 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3ad9a5eca1f4827b972483d6aaad696 : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f5e53b6425294f7387877e1fb9375a2e : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a13ffcdb6ab45b289e1dc22f0257753 : SUCCESS in 7m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a46cc957a6345dda5f7389fec1e3cba : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a410be939034e14a95a630f605d67ea : SUCCESS in 44m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fc3ce9675434ec1b5e346b2a7849b7b : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ef69cd99e61211eb22118fbb10ca3f1de38eb316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 07:52:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3ce0e284c45a7da1da473a8d6813032001b68748","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 08:55:21.000000000","message":"Patch Set 3: Workflow-1\n\nI think we should retains this one, isn\u0027t?","accounts_in_message":[],"_revision_number":3},{"id":"4fdc5c224f25450c58bb771f8e8585128293b529","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 08:57:32.000000000","message":"Patch Set 3:\n\nHowever apart ntc-templates all versions are bugfix version... but I\u0027m not enough confident that we can rely of this kind of info without have a look to landed changes.","accounts_in_message":[],"_revision_number":3},{"id":"4c983c90651dec28eba34651950f9d6576f98103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 09:17:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c365c715d974a4e971f1d20eff827b2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/040c5593bb66406a973903733278c983 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e862d341978244329ff8880030abfc87 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0aaf72a322b4ea686bd865b872f091d : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cc7b33d58814a4588146edee69f1c77 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88502323287940ef8123dccff654954e : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ac6b8d789d54868af41817e7753d3f7 : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e81d8b76efbb418088f198189f39adea : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c34c49bd27684622b1e83b57f1e1db31 : SUCCESS in 10m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f50b5f082e7644d49d726481074272b1 : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90a4937481a44eb3ae3d21fb5f2c5eae : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba80a10833884aa4bf9351e4c2132931 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3192b2862ab945c2a8399a9a66f5faf1 : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3c2a967006147d5be572bdd90d94843 : SUCCESS in 30m 58s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a0b3210efb52421289329ccee5168004 : SUCCESS in 5m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bc7f28798324d01aa19e5bfe7acae12 : SUCCESS in 35m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/edcca7ad00ed497aba98d2f93e6c1737 : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08ba3427f1a649468cd931aad292fe00 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79d210b8162042d1a49487977b7c017a : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d5dbed735cdb4289b2dc75c84a900edc : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9fda9f74fce4151b67ca419f46728fd : SUCCESS in 6m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd6125e6e3c749d289b4364f4c3aabd0 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5694a6a0b4704cdb8e65997483b73fd6 : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d04f260ce5d649eebba02f61d03c4bbc : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6d48131a22c2741ada4c541be937de1081aceeb4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 09:22:34.000000000","message":"Patch Set 3: Code-Review-2\n\nI\u0027m not sure if -W will hold the patch until we discuss a bit more about this topic so I -2","accounts_in_message":[],"_revision_number":3},{"id":"b37828f6eb11596b23ec2d742007c4ae545a6f14","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 16:00:56.000000000","message":"Patch Set 3: Code-Review+2 -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"6bfcb7426fb2b700b8eaab648f4829bfde2573a1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:02:19.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"4346632d4f767bff7167b597265b2b2e03303fe4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:02:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"29f0e80e6d950623c086fcced09516e01fab5ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:04:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9de842eb4a79af2614f661354db1c9fb731e0bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:13:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fccfee6a9e346cabcac5ca7ace58252 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3768f5e98df4f09bd42928b5e0bf279 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac257c321dc34305ae94242e52615449 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d62bd3474642ac9c77c62067a8b3a6 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3eddbef4ac7f4b49b48b75bd1f7c815c : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fa7fc9861b3485abd7e8e542631f643 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/63a7967331924763a13134c175db3d3c : SUCCESS in 11m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/074799fc8e2447d2bee161500d3723ae : SUCCESS in 12m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3907d7a1f856430f9bb9bbec6f8c5da9 : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/860164ddf708439489d41fca8bd8bad2 : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/04234c22f57146c093c18cb96b0dd5eb : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b6a7ed3e0654b06a37ba3a0322563b6 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2d0179bc5804addadf3c6847fdc45e2 : SUCCESS in 14m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ae96b296622649b383e8a7ded05b3788 : SUCCESS in 17m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e1fed8ddc8b34070b0ed5b1a8cddbfe8 : SUCCESS in 6m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4696eed8d8574881b38604028c9255f8 : SUCCESS in 38m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/59b0dd1168e344ed818774ec4381e911 : SUCCESS in 15m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e75122ad258f4f3fab5cb3e5af6e5698 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f86806d7fa8c46fa8141f336d439ab91 : SUCCESS in 20m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/39ff4d82f577446cbb40de25e56801a6 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95b3f00a077f461b987357d80a409ffa : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/27673056b36e404684fee612c854443e : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d16c2648bcf34f5aba611caff6426d41 : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdcb7b85a63141c3b49e933eeca0b46f : SUCCESS in 1h 36m 36s","accounts_in_message":[],"_revision_number":3},{"id":"68785b23d42de4b2ff5c37ad4e886f05af5e7935","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:14:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"79f55caf0c99b9b948c60f69f246329914f841f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:21:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f3243557534cb99bc239107c3cea7b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7735e38f14b60a7333a3d863ea13967fb408c6a","revisions":{"57ddef8d17cb58af6d27b3771e6ec5ec9741dbbd":{"kind":"REWORK","_number":1,"created":"2021-03-11 06:18:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/779907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/779907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/779907/1"}}},"commit":{"parents":[{"commit":"b0c30d1a12b87a4424ed4842486226b08cf7d3f0","subject":"update constraint for tempest to new release 26.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c30d1a12b87a4424ed4842486226b08cf7d3f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 06:18:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 06:18:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Icb07708142f54e89178ca4027b5afabf4dea61a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57ddef8d17cb58af6d27b3771e6ec5ec9741dbbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57ddef8d17cb58af6d27b3771e6ec5ec9741dbbd"}]},"branch":"refs/heads/master"},"f7dfa0f165447e2f209fe871d1998a3e5a127e7a":{"kind":"REWORK","_number":2,"created":"2021-03-12 06:22:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/779907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/779907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/779907/2"}}},"commit":{"parents":[{"commit":"19fba96498eeb97e3030cc0838597c2c2c38f133","subject":"Merge \"update constraint for wsme to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fba96498eeb97e3030cc0838597c2c2c38f133"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 06:22:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 06:22:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Icb07708142f54e89178ca4027b5afabf4dea61a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dfa0f165447e2f209fe871d1998a3e5a127e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dfa0f165447e2f209fe871d1998a3e5a127e7a"}]},"branch":"refs/heads/master"},"c7735e38f14b60a7333a3d863ea13967fb408c6a":{"kind":"REWORK","_number":3,"created":"2021-03-12 06:27:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/07/779907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/779907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/779907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/779907/3"}}},"commit":{"parents":[{"commit":"19fba96498eeb97e3030cc0838597c2c2c38f133","subject":"Merge \"update constraint for wsme to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fba96498eeb97e3030cc0838597c2c2c38f133"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 06:22:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-12 06:27:22.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back tenacity\u003d\u003d\u003d7.0.0\n\nChange-Id: Icb07708142f54e89178ca4027b5afabf4dea61a2\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7735e38f14b60a7333a3d863ea13967fb408c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7735e38f14b60a7333a3d863ea13967fb408c6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
