)]}'
{"id":"openstack%2Frequirements~779966","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I618c48d6cd248fab4a00214b9c42394f9ad518bc","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I618c48d6cd248fab4a00214b9c42394f9ad518bc","subject":"update constraint for python-cinderclient to new release 7.0.1","status":"MERGED","created":"2021-03-11 10:39:20.000000000","updated":"2021-03-12 02:31:00.000000000","submitted":"2021-03-12 02:31:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779966-new-release","meta_rev_id":"0f8ab86ebc72503cf3ac10faaca0f43463bbdd72","_number":779966,"virtual_id_number":779966,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 02:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 10:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 10:40:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 13:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:37:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c24f90bdcd24a046b28806374a769f1d71d037ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 10:39:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e51027da3e35c34b36370f361df631e69ddf45e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 10:40:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c3a6b9d5aa757ac3a8f0b8be437f1eab048cbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 13:54:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b9bae738f5b4da686ad2f1946743036 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44909cbf460c4aa1a549b036d0a84348 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07df192e47e346d1a2554daaaf6fa919 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad9383dba6c4728b5a769c5a796cc1c : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d1b54faae574b8b92f01a3af582f0e9 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f3a51a4061d4dba84ac3b40bbc1eef9 : SUCCESS in 12m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f8d9fb0054f4764ba6cd9059025cd94 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/506c1001d450453b9526e2f648a383cf : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/173e8fbd831946ab999cf5003bb2e80c : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/008f8005ca4c4262b682d2fe9c20df49 : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a7471c5025ef433d9a95c31dd9d86495 : SUCCESS in 12m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2628dba7019e40528e3155226b5a740a : SUCCESS in 28m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7abc653d04d471d87eba17ba77911e9 : SUCCESS in 18m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d260d602d03b4b6997d02e7b797a2924 : SUCCESS in 16m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/818b100d8bdb4f52b3c6135bd0ecd7b6 : SUCCESS in 22m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b113793b3225442cbb893968b0151ee6 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9da933c714024111b4b1b20084d03d92 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f548fb061ac48f088271edad63cf618 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4d5ebe4c0994abeba3323f447cb00ce : SUCCESS in 16m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07febefc8283414ea222c0eab6bdd702 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4f9acb668b34feaa988f2f0311999b0 : SUCCESS in 54m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ccec635efb14df3a0a50c1cbe290aad : SUCCESS in 1h 14m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c50cd8e9c86a4b1ebeb64d78c1ecc494 : SUCCESS in 1h 30m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e3edb582d5c04032bf74199e7215b2e1 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"aa99d704180f1b1c541e95a9e4f11306b15163f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:37:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baa7f122b48fb2ac1076c299fa3f7487cff6dc44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:38:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f680cfca4f9e50455d970b122b67ae3934ed3e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:58:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47ba283b1ee4211bf8d214696c74c70 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bee26cad3b34fad818a7a2649a907af : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de270c08a48247139701a822c7733b1b : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c79194128c6d4c408e09c32edc8d150d : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/563d17944c284d599d6365fe9fd7a03c : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/576e1abbfae249089a1effd901971f84 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a172e0397b4a40f687ce410e93a71aee : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22337ebb32ba44a3a1f6a52c4b56b395 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4659f0ecc3f54b07bfac2fe64e94cd39 : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5765834f2594939a087b204bbe403ed : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e3a74395c6b4dba997a652c70da3b7a : SUCCESS in 15m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4005f73183a94de29972ff93a1d62191 : TIMED_OUT in 41m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b2b460729bd44cb8b177082b5c58b87 : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/29d653a95897489fa0ae50a7c8a7fc8d : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c1b8fb8908f47e293616b742e1798ac : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db9204f3f10d44cabcaf8d1083cad58e : SUCCESS in 6m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/baa123b739d54d9ea3b8ab04c0a985c5 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6cd7f4022174582a878890901e08ab9 : SUCCESS in 6m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/074081c08cd5493c86bca31dcfeaee5f : SUCCESS in 11m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/59569f356b574a13aac27f72f76d434d : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e7250e5e08d4a4588809618f70ca185 : SUCCESS in 59m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/083d91f0e1a147a4b55f37cc6210eb87 : SUCCESS in 1h 24m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3850780f3b6344e9a7ad091cfd80f5a9 : SUCCESS in 1h 37m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/6730f80afe2f46d9a00e606645da78d1 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e69241111a8fb9bf37cdb338d3aa913642680cfc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 21:25:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4369ff0fb9058a21ada5da060b81b4d1355ef6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:15:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7402d5b2e03d4a309d13c10b98794bda : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea007043df74de39da06888dd2b518b : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11c7fcceaeb5418bbea07db02aa0ca22 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ec296be1bfc468b979b86f2c1340024 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6c908f3625341fca2058014b9aa5c59 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e868ec953654abc8e9feefdeba0d260 : SUCCESS in 13m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/24ef294dd3474bada841d68f740c04c4 : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc7baca912c2428db9e413906c0a52b8 : SUCCESS in 7m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26f0fa39fa0c4c48a4bf36f325f1a2e8 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43c32a8760254e1ab92942fe7abf0001 : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e530efa6fabf4c7f9cf382c7d7f66d20 : SUCCESS in 13m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/59d7a5f3c6594102be2104be5ff99b39 : SUCCESS in 29m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d7055e6aca94296a4823738fffe4e12 : SUCCESS in 14m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bc8408141ac142d6a0ee29496b3ddf8c : SUCCESS in 28m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56ac961648a6432e8ca335a315c038a9 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fb0510cbe7aa4f7a856aa3e706962ed7 : SUCCESS in 8m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6560962890a34745a600c742ce9223c9 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2af668f5f9c741858ce6782cac90582a : SUCCESS in 8m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0498b129ac3f470f91082624bd07b590 : SUCCESS in 12m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/795ea47cdf09470ebfcfcd709bf35134 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28383c976b7c450e8acb2631140a53e0 : SUCCESS in 1h 02m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba596ce031d747b29297fa2ec3e3822c : SUCCESS in 1h 28m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e27af4d4026e4ef19fc660a0e4b18556 : SUCCESS in 1h 23m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2bf1b55ede6a4d60a7d4eac6f0799d24 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7258f940890411f89dee36075c3b1d9f4f4f25cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:16:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db2a92ceb7e433cae98836d1fe77e569c8e93f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:31:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a537f5bf62b4a989a3a4cd3c6330792 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa247574b3ba49e9b9746d0e824750e1 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/551062ec802f42779dbdc8288b0b6a7b : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/672aa744735244dda9b2477581efecfd : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd4630ee6c7f420b9bddc4197f0504d5 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f61e9abc73c4e2f9d55062c6abf5308 : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e1d47e259d54247a5bdc5bf21ae26be : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/93b62333b4de45c2a22d9702b13ddf98 : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eb065e71793645009ba8ea8060a4feb1 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45079a2ad79b4605b6dafd4da5061a86 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1fc52d96d644176ba40d935518d37df : SUCCESS in 12m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6c9aa27b2380413484238ebc018e06d3 : SUCCESS in 26m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5840d159c94c4a1fb37b8f941a78a450 : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/30f51c6e764f45aca7251cca937be244 : SUCCESS in 16m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08afbe2eeea347ddb2ce8acacd10da21 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f4c6f25a40441b1b20194bfbcd19619 : SUCCESS in 5m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9a3894e42954477b8b0f758d188e1ef0 : SUCCESS in 14m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d56ff63e3f44ca98d70ffc31d013d04 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3976315382934dc5a3998bb127817ea2 : SUCCESS in 11m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/98ebdb8bbf104c019266567ed6581ffe : SUCCESS in 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/709838cd55f24bb6891175fe4d6f0747 : SUCCESS in 1h 02m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b291b46f387340c7b8ee01effa1265a7 : SUCCESS in 1h 23m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e93feb01d6c54fbe858e2cc2ebb6e75e : SUCCESS in 1h 28m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/d9324b4036bb4e1db6af8c9a31642889 : SUCCESS in 1h 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0f8ab86ebc72503cf3ac10faaca0f43463bbdd72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:31:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de3eefc409ed87321d91cb1d6d0d3ba81d0fadf0","revisions":{"de3eefc409ed87321d91cb1d6d0d3ba81d0fadf0":{"kind":"REWORK","_number":1,"created":"2021-03-11 10:39:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/779966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/779966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/779966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/779966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/779966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/779966/1"}}},"commit":{"parents":[{"commit":"3cf902b6000525f3d8787ed045ec3625bf2c498c","subject":"update constraint for oslo.cache to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cf902b6000525f3d8787ed045ec3625bf2c498c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:39:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:39:16.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 7.0.1","message":"update constraint for python-cinderclient to new release 7.0.1\n\nmeta: version: 7.0.1\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Iabaa074147dc38125e22dd8f3ecb8ea664bb3bd7\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I618c48d6cd248fab4a00214b9c42394f9ad518bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3eefc409ed87321d91cb1d6d0d3ba81d0fadf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3eefc409ed87321d91cb1d6d0d3ba81d0fadf0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
