)]}'
{"id":"openstack%2Frequirements~779967","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I3231e76ea21709287e30f1f6eb1820682044a673","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I3231e76ea21709287e30f1f6eb1820682044a673","subject":"update constraint for os-brick to new release 3.0.5","status":"MERGED","created":"2021-03-11 10:39:53.000000000","updated":"2021-03-12 02:30:57.000000000","submitted":"2021-03-12 02:30:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779967-new-release","meta_rev_id":"5e335d9d3c7758170c13451026315359a3941438","_number":779967,"virtual_id_number":779967,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 02:30:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 13:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:39:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5514265f043a6e15edde392ab21f0ffd874b8fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 10:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf5d33e01716e64f212f98a90efe3e9f22f8ca0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 13:45:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c2f8f22d7b42bab34e0729143b12ae : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0df968e44f0242ff84b4f05396e92352 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efdec0dd8ffb466d8d2e83d9b54cde16 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8fbf40614b142bab518e21ba2c8adb1 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85018236813048629a1cb14c143f5aae : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e983c5b5ae6c4e9ebae668b21a07085a : SUCCESS in 13m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/19780e0a5aef4ee39d3c1cf76c0f9e2f : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9df41b0bf9484f729adc982bdd101c0f : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/511af81f991c4de2b1dc6d174e06c35d : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a33f907853a4364ba90c28b0223745d : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b804ab33ae3428b9847f117708e4a35 : SUCCESS in 15m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f2a01e6a2024c14b6453f6072777093 : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f22e07abbea84b328ed8c247fe164da4 : SUCCESS in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/62661892556b4b209db91e4b45ae94fe : SUCCESS in 16m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c4657f8ef004674ac9bb59d74496fa6 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9dee072c261f43d995ebe9046ee0e38a : SUCCESS in 5m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/25af3bddf00f40bd9ef5df960115dc14 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5a74ceadc8a4cc69bb26846be67b2a3 : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c488cedbc4e4cfcb7e13024cf8c6fc4 : SUCCESS in 13m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbcc846f06bc4b5aa7f417ec8978ce17 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e26c45c421b4464838a95a3d1b21329 : SUCCESS in 1h 04m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebb556841f8641c59d51bb3b228ac0f5 : SUCCESS in 1h 16m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b27f1dbc85904c128dd6b8fa6a3bde31 : SUCCESS in 1h 16m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/c10ec19a6d8c4ec1b66106b8bc75962f : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"20b039b09b68096120de440b5e05e8a9b8891606","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4237016c432f93ad9446282d8ac5ac607dca4445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:40:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a62e64fe18281a978d646b171a65060752417b2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:56:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99d31104a0d474eaba5ea4563f31a92 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a34b10bf7d435091bf13007138f6f1 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2821c5b2dd843548d9bde77b71ffaf1 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9896725397754426b6233caaa2dec44a : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ec2b4ff780a4bfd9c569ed574b09c9a : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7236ad0a5b543c5bb2665079237f637 : SUCCESS in 17m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cea4f427805d49c99516cb84a01365f2 : SUCCESS in 10m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b55e4d92871a42c587f791d55250b993 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/50f5d14480ad4462ba473418a144d0a2 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ce0327f18dc4048859c16cf13dd73be : SUCCESS in 10m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c28b2fc468a4e92b4c70301694fa151 : SUCCESS in 16m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/af2c50c2ce1445d0904f74bc0567c1b0 : SUCCESS in 25m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0ef2065301934fc79871d35f2ccf8aaf : SUCCESS in 18m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1612c59eadb7412a878d13fca50f7265 : SUCCESS in 16m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e226f96ae40c468bb53223771eb858fc : SUCCESS in 20m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/70df0f39754d40e6b8c9ddfe15a99638 : SUCCESS in 6m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9f8172ecffd54d58bf173476470d351c : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8dc69b4384340188b687fe81bdb07d5 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b33e2e1f3c34626ab55bb71d5903e7e : SUCCESS in 12m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d78d522f2d704b63836734f1540835ed : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a95c59e764b74f5a91612dee964d4dc1 : FAILURE in 57m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf20df61fc584ceabdcb6ca48488b7f7 : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d1646cb26c24ea48734d0daf0b6f1f5 : SUCCESS in 1h 17m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/dde1582a336548cf854fda99488ac9e5 : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e06e6ef9ddcadfe753ea625f944d714877957fb2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 21:25:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b33f55acb9a50598bb423796a9f7b83212d5548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:10:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f920973f044756b03ed0e9dc1edbe9 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/265f227f846143eb829148191968f636 : SUCCESS in 9m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/194d2bd8bc86436b894fda8c89121c15 : SUCCESS in 8m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc0fb9b5badd423293c15e778faf5e44 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cb7521ebdfb412fa59ddaead60c4193 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ab146c6dc760436592126d1b32ab7828 : SUCCESS in 12m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/810d8e2c1a684883a7de14defa41b4eb : SUCCESS in 10m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ed3d029c0fe4d4c8891efc246ba9921 : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c7a36862fd5c4f419050453fb0cdf719 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb8b11364e3742e0847831041b38c412 : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4cb47aac4e5648f0bc7079cb0864628b : SUCCESS in 28m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3529f430694246bf8ef90e132856f0ae : SUCCESS in 30m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea562a0908504b50bb4a8e2b1bc5b2a9 : SUCCESS in 24m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3e1854522bd74ad9a0d01101613c0d70 : SUCCESS in 15m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/222f5e08816c495d866080eef193a822 : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8d7f1fb944c4aeeab37e776acb775b2 : SUCCESS in 5m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0009013fe28a40a097772fbae00433eb : SUCCESS in 8m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a133864a7b2e47819f6f2049927a2153 : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a89377db65fd4d55843b7b036931a299 : SUCCESS in 10m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87367462b15145199abfeab5d224c5bc : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd1d5d8b86954e2cbecc639ead1e6944 : SUCCESS in 53m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/237a7003f23a4db4a94e740ebbb08a2f : SUCCESS in 1h 15m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa078ed3b28e46038bc92cff275d9e4d : SUCCESS in 1h 25m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/9d0381841bbd47e6822fbe2e5406df12 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":1},{"id":"252fc0177fd18177f580ff9e7e5ded5dee1cfd4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:10:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9affad36fda96f1b158ebb99d3fb21f4e8da58e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:30:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbbf191ee34b4aefa615acef192546f8 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d448ae0400848b1a66f2e0dde62c18b : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1d1dcc1e129456498577ed222413d12 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9da08c53826c4c14a2e6c01c272a2392 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7c398fce02d463b95740bca930ecb6e : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72d48c25ea33480d88aef2a1478ff2d2 : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/441013c19ef74b978fbff5f6d07950b7 : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d1ed8be5e6e4da4b2c15ded7533cc5b : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a9bad8888a6e4c2e827999a372314d0f : SUCCESS in 7m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f720cac45f984d16afc61fbc995d4a23 : SUCCESS in 6m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6c4d0bad19754089be144b3e790feadc : SUCCESS in 24m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dd1301d45ad847fa999b00899612e1c8 : SUCCESS in 27m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b52d33ad61fc4ea788e4be23bc092b6d : SUCCESS in 14m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e0c71186fa7147c0b406ee37555a442d : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc1ad26821954e78b976e59adfd8c4e6 : SUCCESS in 13m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7397af06f2fc45b585268820654d395f : SUCCESS in 4m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bd68f46db53b4bb69f8b2976d698feb2 : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd5cbc459662435486f02e5f281088e9 : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a2feede61ef94018a3a8b381c42465d4 : SUCCESS in 19m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/579c525b2b0a430389a214c958cca554 : SUCCESS in 19m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7960c2fe2f0548d5aca5d55068024e0a : SUCCESS in 47m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f0d31adf7b64bfa998a254ff38be458 : SUCCESS in 1h 02m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b7aaeae279e48ad99e91815ef4d5a20 : SUCCESS in 1h 46m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c254842e768e45019f2f8de4890bfce8 : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5e335d9d3c7758170c13451026315359a3941438","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:30:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f81555b752e6f1c8a77c34ccc10a842f2b6928f6","revisions":{"f81555b752e6f1c8a77c34ccc10a842f2b6928f6":{"kind":"REWORK","_number":1,"created":"2021-03-11 10:39:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/779967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/779967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/779967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/779967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/779967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/779967/1"}}},"commit":{"parents":[{"commit":"3cf902b6000525f3d8787ed045ec3625bf2c498c","subject":"update constraint for oslo.cache to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cf902b6000525f3d8787ed045ec3625bf2c498c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:39:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:39:50.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.0.5","message":"update constraint for os-brick to new release 3.0.5\n\nmeta: version: 3.0.5\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Iabaa074147dc38125e22dd8f3ecb8ea664bb3bd7\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3231e76ea21709287e30f1f6eb1820682044a673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f81555b752e6f1c8a77c34ccc10a842f2b6928f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f81555b752e6f1c8a77c34ccc10a842f2b6928f6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
