)]}'
{"id":"openstack%2Frequirements~779968","triplet_id":"openstack%2Frequirements~master~I3767550d85d74f4ef2716c6dbfa06f4a179f7690","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3767550d85d74f4ef2716c6dbfa06f4a179f7690","subject":"update constraint for python-glanceclient to new release 3.3.0","status":"MERGED","created":"2021-03-11 10:40:37.000000000","updated":"2021-03-11 19:20:10.000000000","submitted":"2021-03-11 19:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779968-new-release","meta_rev_id":"ea268b97c2176474e8846dcbe5a80fa2a39bca95","_number":779968,"virtual_id_number":779968,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 19:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 10:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 10:41:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 11:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 13:58:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:37:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7dbde23bfcfd63e076cc94fb055a8fa2f59b5539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 10:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed7d8c66986a61c82ce8edc7194be263115b2611","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 10:41:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cfa9e073da4a0e09047a6dd1e12e998d54d21d0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:14:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e132a4c17b82481b8213f78b87377c31 : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1ba5d7998967705beef3c9c72741d8fd5c78e075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 13:58:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d578b4337449d58fe8406c7c7105a6 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a201d7f3dc483b91cd70ffa7d57568 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce78029a91b43efb6e0768357ddbd3a : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40537b454a444474be66e7a99d541269 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13cf3b959d334456b3498f51e7753b30 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c3e672ab6ea4276b811d3dae5a86223 : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0117c99c81964a17a6f9cb2c22db2e02 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53a5304685e0424a8a01b53e671553f3 : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58ae160e0d73460fa1dd15084ab25c94 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5a229f3d51114264af569bf39901d118 : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e728df68667a4721b5f497630462c701 : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76cfa1c3fe5e48c48e4937df494403dc : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f0d4b093b9d49cab5053dff11ae893d : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ce75d7e1f3f4c7d8d0414acef58a165 : SUCCESS in 15m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c7d5a4c013d74571b156d475f37cc42a : SUCCESS in 19m 31s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/cf87b25c8c7f42eba7a4a3ff60bf979a : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d9c745dc9ab402595da3e225547aaab : SUCCESS in 41m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b6673398cfb40eeb4de3c70c9a71045 : SUCCESS in 20m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3129b303af7c45308d3cdd09a0bef29a : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51e63ec301854d29a33eedaeb4cbabb6 : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b09dbc2f514452b8b46fa6bb5edd7f3 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e455ca0e87d4447aa460f116c5a88f15 : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/af1f17bde7604dff9a02d300b40477ff : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87d43d7dee5947a1be6c7054d6a6e794 : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adb3a876fa844b3d8b95de59e03a464d : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3aa703aace39a51043d08520fbc91ea282f5c3bb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:37:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8e206b0086b728e25756f6a193c6af98eba5f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:39:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"239acf2e784ee03471427ec33d5eed8d705c0824","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c479c435734010b1c8387d273b1829 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b43d5d1523482dbcc8e358886e781a : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f04fef7dd36d49f98465b146a5ee074f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175cb73f396744569faddfca762a7919 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7a665e06612474aab00674335d7d6ac : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3cd35d48b96b4c2e84c86b19ee95a96d : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ecc27d9fbddd4fef99a49ef49d7a0933 : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4a236d8f1a24e468502d594f1e0c92b : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d11c04392cef491ba2c936f2282cb865 : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/58830e3c608c47f99af7256c328ab87d : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe919d0dba994d75aa1712818244db27 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90cbe5bbc80e47e0918d0188afb74755 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2036ba0becd04082acb2dfa00e3f3d65 : SUCCESS in 9m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e22030b79f064d0d841863ea9b2939fa : SUCCESS in 15m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1ff30cbe57cb4c719fa4b4d19060abeb : SUCCESS in 6m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c3de3b7790f443eba0de30a308109f2b : SUCCESS in 38m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/851b3827f116455a8b1341353cf6810f : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e70776ddd144a29a50069001a7821b8 : SUCCESS in 6m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ef62e95c29244838f13c6c16aa6fd4e : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/52136ff84a6744e4a94c63775d279663 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86984a78640a4f4ca1c0a331e27194e5 : SUCCESS in 7m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e710ab1cf13c4cda9909c35d7c2d2fdb : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ac00fc31bf24e5baa17432a3264ea59 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d730a5cf399249cab845846cd8c06852 : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"276bd6e9ffdda011af2408b23101d702fb93df87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:18:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea268b97c2176474e8846dcbe5a80fa2a39bca95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:20:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5523182fcd4640b986822d6f7184fc : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b9dce86243a21af8124560b8bcf5b2db5a478bb","revisions":{"2b9dce86243a21af8124560b8bcf5b2db5a478bb":{"kind":"REWORK","_number":1,"created":"2021-03-11 10:40:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/779968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/779968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/779968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/779968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/779968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/779968/1"}}},"commit":{"parents":[{"commit":"2f7d0146f4972ad85fa7426a2084414d5efaa38e","subject":"Merge \"update constraint for python-swiftclient to new release 3.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f7d0146f4972ad85fa7426a2084414d5efaa38e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:40:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 10:40:34.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 3.3.0","message":"update constraint for python-glanceclient to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Change-Id: I0961b300915a21b1792d771498220e1b9760719a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3767550d85d74f4ef2716c6dbfa06f4a179f7690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b9dce86243a21af8124560b8bcf5b2db5a478bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b9dce86243a21af8124560b8bcf5b2db5a478bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
