)]}'
{"id":"openstack%2Frequirements~779983","triplet_id":"openstack%2Frequirements~master~If16c2a6fe85cd2e71e0f6e1b29d61ae184800243","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If16c2a6fe85cd2e71e0f6e1b29d61ae184800243","subject":"update constraint for python-senlinclient to new release 2.2.1","status":"MERGED","created":"2021-03-11 11:12:03.000000000","updated":"2021-03-11 19:21:25.000000000","submitted":"2021-03-11 19:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779983-new-release","meta_rev_id":"cd855b24b70b483bd88efb13bd758b266782f083","_number":779983,"virtual_id_number":779983,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 19:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 12:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:37:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 11:49:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 12:20:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 13:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:37:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"56ce94b4d751eb13d7ba5649dbe3558d0d2b5089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 11:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c292eb9c7a440bedabadc21aa9e057eae3ceaafb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:49:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/08523dc6ad85443f9fbaeea752693d7b : FAILURE in 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5408beb1a62f2fdff242e9a994d57cc4b3500bcd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 12:20:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"371d156297b3cfbd96ffdfdd150d5fd8f92458ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 13:58:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb7b31ac77346219834536fe93cffb6 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92f78d2ba2c4e11a03aed8e910d0206 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a24fb5baa46344ed8317437dbe821f57 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69da3db23b3e4336bdc8485b3bd7f09e : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56199a41dfb45d28c6ccea1456f7fd7 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ded4f6538fc4d08a63a68686c985858 : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88e19f7e172f48c9b0fb629f10b907ce : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e02e2ab317854db08138c25860c6cfd3 : SUCCESS in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb5f2c4954e748339634d1de18d57e34 : SUCCESS in 11m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc2c3280768b43378cf3e40685ad6ab4 : SUCCESS in 12m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/55b4f20b3b0642f28c8b8cbaba16c135 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b2e8fde04a474dcf95deea9ce253a08e : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cdd8686fd1a47b5a9c46a100a1feefb : SUCCESS in 12m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5fe22197abd84bcd80ee6b6e906694c6 : SUCCESS in 12m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bcc6d3e729bd48b99ff36b4aa0eac63c : SUCCESS in 17m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/79d377be3a48499895657d07d950ae91 : SUCCESS in 5m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7c420fb427e464f80ae8a7790614019 : SUCCESS in 36m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9df79e49e90492e8757414c9e692393 : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65f0d2d9c8524633b0b53039d9cca268 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2f5694e02e04bd38b8206255c6236e3 : SUCCESS in 19m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c87a24a4aac4d47935cbb8bec7750ba : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/34753785c57f4084a9c0cf9ce8a7cf74 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b9d4dc5e6bcd43b1946c17a0168105f9 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5371ae49ce324b1ea42abc40e486bf54 : SUCCESS in 51m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e8252b3aa634737aedc141662095c10 : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"88de98977e121774e516c9585ee1806604814aa8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:37:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07edee900ba5e84e361722eecde0791342ea8366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:39:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bf58d44f44f3e851cff2b8df1decb0835a0ab71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:18:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8beee05926674f6ea2b18862ea0bffad : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d269f3d5294af3b7cfff4ad5fcbc86 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f0a3025a7c4c74965a067a10ef175a : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d2b60d45c64edc8d635780d42c6263 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5577a939b0f9411892eddb84934bf010 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2994c768890043339d644dd9093667da : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2bd5b93522e41a484d65964ddf1f09a : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d02c6d8266a4145b9098fba0fa9c79d : SUCCESS in 18m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d6a441598bee4773b8a3a872c00ab0a2 : SUCCESS in 18m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/968edbff82af4e14a4af883fd3a6b7d8 : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f163a5c8c3c94198a9bc3e26692befa7 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c3b2f5f407a4152aa1fda24fe440f8d : SUCCESS in 11m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/93a17822ab9d454182ac86e340e2a547 : SUCCESS in 12m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/13ca6f2e8afb4b91b61a5669083516d9 : SUCCESS in 15m 53s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/99b2017e1b1c43b198632ff425f546a4 : SUCCESS in 7m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd408e1783d0418aaf8c6f6f23421b6d : SUCCESS in 38m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fa0da7a9fabe485ebec4cdeb06de7622 : SUCCESS in 14m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66f52bb891dc45a891a587f30539ac6c : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1751853c0d34c968cdd2101046e06e1 : SUCCESS in 27m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f6bc63b88a884ab88f89ce106ca7e375 : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a97261d60b44d7ea6214f7e270b6919 : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9fab19a9750647c09860ccaa9f628ceb : SUCCESS in 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2359857cf8924d19a6f77aaa66ad8a88 : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf5edb038448465ba6e0f566b0696e8d : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a9f8c5e52b5da75a1ef02ee8d73aa3d36273fd92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:18:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd855b24b70b483bd88efb13bd758b266782f083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:21:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b556deafe1f34d7cb6fe09d276f7639d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17a9c7cb5ace53e104983aec2825b5991e9198aa","revisions":{"17a9c7cb5ace53e104983aec2825b5991e9198aa":{"kind":"REWORK","_number":1,"created":"2021-03-11 11:12:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/779983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/779983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/779983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/779983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/779983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/779983/1"}}},"commit":{"parents":[{"commit":"2f7d0146f4972ad85fa7426a2084414d5efaa38e","subject":"Merge \"update constraint for python-swiftclient to new release 3.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f7d0146f4972ad85fa7426a2084414d5efaa38e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:12:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:12:00.000000000","tz":0},"subject":"update constraint for python-senlinclient to new release 2.2.1","message":"update constraint for python-senlinclient to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Icaadc146f5608e0de7bdab672a3fd4fa4858b028\nmeta: release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If16c2a6fe85cd2e71e0f6e1b29d61ae184800243\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17a9c7cb5ace53e104983aec2825b5991e9198aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17a9c7cb5ace53e104983aec2825b5991e9198aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
