)]}'
{"id":"openstack%2Frequirements~779984","triplet_id":"openstack%2Frequirements~master~I85b26379a83890de05f9fcbffae17d97566803e8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I85b26379a83890de05f9fcbffae17d97566803e8","subject":"update constraint for python-octaviaclient to new release 2.3.0","status":"MERGED","created":"2021-03-11 11:15:36.000000000","updated":"2021-03-11 20:01:05.000000000","submitted":"2021-03-11 19:56:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779984-new-release","meta_rev_id":"8bb6015f4b18428e50ccf7994b874d85d929ad8b","_number":779984,"virtual_id_number":779984,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 19:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 12:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:37:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 11:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 12:24:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 14:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:37:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c670afcd357ea43d07f678149ed48bc799f2ee4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 11:15:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2a9bf195601584b034bfaa962fac2fec8fb300b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:49:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17c0a742fcec4615967cfcd1712b7603 : FAILURE in 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"bb4e71a7c30b553db9227ca61859fd6fb3a423db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 12:24:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c021344992a756c1113bde5009343e7a1c18b1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 14:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3363d0cee043448f95b9fd6cd2ac5806 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be9052081cb428f9be97606fe90a08e : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa1dcf817484874a2a5b9b930d7d3e4 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff19c0cd17f483595f36925898bb4de : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a006453ad1d4fb99feb544e90b5a45b : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3096f86b62440d28f47886117bfe1c5 : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9dbc16b9b2984b4daa86fae99b5c967b : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cd2e34f1935e4624846ce99998e1c946 : SUCCESS in 10m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae7f51362e304f91b97df22f90004b48 : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e252f21cf1e045cebd491bb05b8291fd : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6239601dd42441449dd154319153c557 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/955fe853fa3546d9a9d5d04b67861010 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27ac05a7544e40a291a0565e1c7eaafb : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/847bef7bde184dfd91011ddbd654b025 : SUCCESS in 10m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8372fe19e2ae498ba31f4152feccc31e : SUCCESS in 14m 47s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b9c3498666b641ae833bc01f7d6bced5 : SUCCESS in 6m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8af08d76bd5d44a9bd9e43a64ff5a464 : SUCCESS in 37m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/607049aa719f4dc3836a024387fdc5f5 : SUCCESS in 19m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f8e00015cca4c268f930c86efac713d : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2179cf5d3f454710ba008fe2dc1939fe : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f4479549803d4790a4fa9c9a6dfb1eb2 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0533bdb038b446f983ed04972c7135ec : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b8701f773a54300903184e05d8f08f3 : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64566cd1b6974b28a2c0f1ecf8b7defb : SUCCESS in 1h 02m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e97a842ba134adcb8dfdd3d175bd716 : SUCCESS in 1h 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a1d0bf21333dfd1b3b107ea25e38eaa9f1d1145a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:37:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"106656cc62e5430f70503c12c48105e1b257051d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:39:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5752971d73b1d6278685fec6be27208bf19d97d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:56:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c347598d3144809c8ef93adf70a2d1 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3d86174b9d424aa93a95ee6266112d : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8f5c7c8311346f086c7f2e649cfea4d : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dccf349338414928be9e1b4efa60fb6a : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c70f98ced03545f88325fe2ae2bfeb94 : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/495f362b8301406a8e56da915334fa6b : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f093b083551d4d9b9f57f7dd22a2cd88 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/88ff4fed997b4876bb9047b5e3162ea6 : SUCCESS in 11m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0af18e02058144f3b31a87b96f3fa080 : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c9cf44e1fcd4cde94c6e173cdc98af3 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a57524b47be42a5a375b8ccc367bd41 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63d2ba25724548c9bd394a429fb25c4a : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c488479e6a04d9c826473b7e71b2580 : SUCCESS in 13m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/433f626d5d7448afbae4b9056bd4940f : SUCCESS in 14m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/25f0df42ba8b4ef48f5f03360dc45452 : SUCCESS in 6m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/874bca619a6940f08391dd63731ccbb4 : SUCCESS in 38m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/656b414402ad48f98fb31ab42896866c : SUCCESS in 14m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/649e82613bc24f31a3e8b4cb510ede7c : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dc9e8bd45344f1c9c2b012b80208c79 : SUCCESS in 19m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d67a3335c1174e44867fda3d37e1663a : SUCCESS in 8m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/243536b4ab494efc8a7e5c8c7b40f5a0 : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f1fd94986c5840449ca001a7606aed83 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ceff1ff781641cbb1cf30ed29584381 : SUCCESS in 45m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/703bf684a68e4cf6bccebb0a4155e2f7 : SUCCESS in 2h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a32e88a0706d33415b967baa173cae83055c7bb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:56:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8bb6015f4b18428e50ccf7994b874d85d929ad8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 20:01:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdaf04b6e8f4be5b458cfeeeccdcfdf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d900bed0b09e24f6db31ca795e43f3d4e734a50","revisions":{"7d900bed0b09e24f6db31ca795e43f3d4e734a50":{"kind":"REWORK","_number":1,"created":"2021-03-11 11:15:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/779984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/779984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/779984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/779984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/779984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/779984/1"}}},"commit":{"parents":[{"commit":"2f7d0146f4972ad85fa7426a2084414d5efaa38e","subject":"Merge \"update constraint for python-swiftclient to new release 3.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f7d0146f4972ad85fa7426a2084414d5efaa38e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:15:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:15:34.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.3.0","message":"update constraint for python-octaviaclient to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: I145846fb5c4dbd33c3f28700bee9d8f05f7296f2\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta: release:Code-Review+1: foundjem \u003cfoundjem@ieee.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I85b26379a83890de05f9fcbffae17d97566803e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d900bed0b09e24f6db31ca795e43f3d4e734a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d900bed0b09e24f6db31ca795e43f3d4e734a50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
