)]}'
{"id":"openstack%2Frequirements~779985","triplet_id":"openstack%2Frequirements~master~Icbe42db3760193ab1ffa7975a89ac0a8c3a7e975","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icbe42db3760193ab1ffa7975a89ac0a8c3a7e975","subject":"update constraint for python-monascaclient to new release 2.3.0","status":"MERGED","created":"2021-03-11 11:19:16.000000000","updated":"2021-03-11 20:02:52.000000000","submitted":"2021-03-11 19:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779985-new-release","meta_rev_id":"8231ecb8d612ecf3737ff7a0ff29fa2348096e46","_number":779985,"virtual_id_number":779985,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 19:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 16:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 12:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 16:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 11:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 12:25:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 14:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 16:37:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"17669a231c1e09d084d97268f51ca6c0baa541ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 11:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c1ac6dd803039fb7666b096cd5f5ec366ede04e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:56:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/792c609f19ad48cc83ae1d391ed65375 : FAILURE in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"31b7b4521521b5f721f31f14f22f66c99b9fab00","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 12:25:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8dabe121415f425340a96444636c37af28362e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 14:09:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995b951676f24d528f770c82125fdca0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0cc194e8c64760a05e0a7eaf9619db : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def3d2d3360b4affbcfb9f9cf8e49ca6 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd476e64b5b46a798aca78f647adfdd : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b9f01c6310b47dab10ff367eb553b36 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e87a413cf4b457897ccb2d60869de3b : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72e0569df42149bb8503ba0b1703c85f : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69fe7f58c3ed404e836d9f6c9eb65f67 : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/078f85b9ebde49b18b615ccde83f6b6d : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7258131e6542457089c1b4c95c753b24 : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/34a2000b939e427186f3b3082fbf9f57 : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d656919f86bb4731814eee16f7dcb819 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bc6c35bf0c746c0860b6c2a7b8e98a6 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ee84eb314484befba000a1770b0c103 : SUCCESS in 11m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f503f72a42694d5297a9f35086279147 : SUCCESS in 16m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/83563c81639143cab1d8b12dc0e06669 : SUCCESS in 6m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/38b106f9e64c46a487a7f490a8e8cd54 : SUCCESS in 37m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c65e144c7404df7bb908ad03e7228b1 : SUCCESS in 15m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1366e17d70044868872222609d10908 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b404736a85543dfa6bdb642e9654741 : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/94c231cbd6f4420a81b4f301e8f29bd1 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f27cac63a8e477a8de11d4124db0d9c : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72d7c13961c041bc937cca2cd634dfc9 : SUCCESS in 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2af2e3b9cb24e93bfebfbce1a1b3b87 : SUCCESS in 1h 11m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13c141c43ab84210aab33cc76b4eee12 : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cc52903728879cc64bd08b8f3d9753fd5ced9cf8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 16:37:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"deb9c0b732836f4cdb2cb3a269e298b1a2c52f92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:39:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37c78329f5747357e20606edf11f857311a9f272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:56:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf1bee016c64f669d45b8ef12a1fef5 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab6416d619d4a9aa2a046085d9b2e9d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09374de59d094330a1923bdeb566cdd2 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603b93060ed34a86856919e1002c811b : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d40ea2bc7e04cc6a07d926916a81b55 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67d7992fa3214da7b2065662b97469b4 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b9a508e8adb45a281ac5bb840cd7f57 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7156ea60264847f68823971adc261e59 : SUCCESS in 12m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/73a953fc46904458bb9c856db46b5ede : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/10f6fec412a34e1cbe5a15b832685e32 : SUCCESS in 8m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/915bce1b2aa44b298a316751af6f6643 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95fd477b97294c16a382aa54c790dfdc : SUCCESS in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73c08357c11b4c4686eb8b3f1f5375cf : SUCCESS in 14m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b516666856274a0f948692f84514594c : SUCCESS in 16m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b6d6e0ddf17d454b91a5359286857b54 : SUCCESS in 6m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/357a99fc38a6425e92c89b39278b3a41 : SUCCESS in 31m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/74e3332b12804db4b3b26101888a4c51 : SUCCESS in 12m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba275ffc14274ba79c952c1642953f21 : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d73f9b4f3e54eb59149c74e0b9368a1 : SUCCESS in 20m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/466c6198a62d415193158674266f490d : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f3ecf0c7c62242f8a7913fe9523dfdb5 : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4fcf7e9c36264ffd95ea6429243b3857 : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e65a463d52f48d59630506eaaf9e224 : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12fd719d2c784bdcae923927c341dea4 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"00a87586279d521934d569f1b954c98f1966265a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:56:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8231ecb8d612ecf3737ff7a0ff29fa2348096e46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 20:02:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38377a5b4ddc4e259fdf05e6a3f0468b : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf2770c9e4c6dcb75d87912fc68a107e71410d28","revisions":{"cf2770c9e4c6dcb75d87912fc68a107e71410d28":{"kind":"REWORK","_number":1,"created":"2021-03-11 11:19:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/779985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/779985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/779985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/779985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/779985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/779985/1"}}},"commit":{"parents":[{"commit":"2f7d0146f4972ad85fa7426a2084414d5efaa38e","subject":"Merge \"update constraint for python-swiftclient to new release 3.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f7d0146f4972ad85fa7426a2084414d5efaa38e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:19:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 11:19:14.000000000","tz":0},"subject":"update constraint for python-monascaclient to new release 2.3.0","message":"update constraint for python-monascaclient to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: Ib1e04d375b1f42b1ab74edca6ab350c1ca174be5\nmeta: release:Code-Review+1: Martin Chacon Piza \u003cmartin@chaconpiza.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Icbe42db3760193ab1ffa7975a89ac0a8c3a7e975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf2770c9e4c6dcb75d87912fc68a107e71410d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf2770c9e4c6dcb75d87912fc68a107e71410d28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
