)]}'
{"id":"openstack%2Frequirements~780102","triplet_id":"openstack%2Frequirements~master~If6665443854e72e02a3da91eeb8815a117fc1a15","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If6665443854e72e02a3da91eeb8815a117fc1a15","subject":"update constraint for python-zunclient to new release 4.2.0","status":"MERGED","created":"2021-03-11 20:11:19.000000000","updated":"2021-03-12 00:46:50.000000000","submitted":"2021-03-12 00:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780102-new-release","meta_rev_id":"e7d221f86371155117a769566297789c71b0a435","_number":780102,"virtual_id_number":780102,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 00:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 21:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 21:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 21:25:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 20:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 21:04:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 21:25:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-11 22:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1aa06e053f906e2d75eb3e30419a4d4a08dc613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 20:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f16bd27a193fa9f0aa2520b6ec07fcc2e172bb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 20:56:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c7047f851d48426f945af0952552fc21 : FAILURE in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"044e70040b3d166120cd28abab059ef720d4b36b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 21:04:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a473c9405c07403ffccad3a84bb3053c0ed8fa51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 21:25:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d07dfda93e99be0039b6f9d6a6807594fe4c7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:23:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d701a20aced640e9a899c851b78f0ada : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42d60cc86684ac899d8a4172efe48a2 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe1c30098624285a9954a03025b1396 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e3680dbfa14d6bb634daf85c2593cc : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cce584ede8dd46c0bddab752af90e0ca : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b25afefc190f498c895ca032e0dc6db8 : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd3d5e4569614959b8c2eecb313338ea : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a40a87fac04042a09ed8cc7f1efce841 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef172d53b66341c5a8bd52ef9fbb3a63 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0bd216c170934cdf9f011054751e6c3b : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea718b86e0fa4900b90b960cd545951c : SUCCESS in 6m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4303b90d57194931a8aa4d55dffc7d90 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/908c69a278fe4356afc92548818e9257 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/57f8c7f5a47c46a5831dd60e4760b26a : SUCCESS in 12m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5fbf6e2123db428c81a4cd368071ba07 : SUCCESS in 14m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/885ecb1c597a4a95a3a4ea020463cecb : SUCCESS in 6m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1aed0f1f7704444b82cb1d80ad3179a : SUCCESS in 39m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d327aa2552154c109ac3a6572a99ec83 : SUCCESS in 13m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eddb40e998f64dce8889347672e74db1 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab406dc59c234673929f5122f4ce4d43 : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b0bab00daea40c8a2e43b71b1352f8d : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/674bfad119744fd4820d9a6cc5b7bb49 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e9478cfcb7f46d5a1d1c63a87ff174e : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/427518d9f32b4cabbf8e616cbae0f9f9 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22bc834b027f415e8d5ccbac9dc6d633 : SUCCESS in 1h 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1e5fe5719b89c0ebe733887b5710fdb928bc88ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:27:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cca572b9eaa849626c89ab81d7f641ce0ecd2ad1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:45:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0f64858ef6401a8920b3f12d0bff81 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d0a374a8e0496e9fdae720ff4593d7 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be7f5beaed274fa4a59b413de00ccd00 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c483ef945a294be5b6ddd54bd083a066 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8bf446e9440e445aba1ac75162c96f87 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fc8b4d6df1741dfb6e699112989094a : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/61fd724581c94481b013955f734e0bc8 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7ea0fb3b742e40128c2d683ff2c0d004 : SUCCESS in 11m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b134bddcb5804b438d3d0de9ed71e6d2 : SUCCESS in 18m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/437e29b2513749378c54b2cd2a7fdcc8 : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4fca3d4774d4dbe94302a0434abf2fc : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/666e09b2ad334727a437a36086e084e6 : SUCCESS in 9m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a00689e7bcc64596a668e853d15116c4 : SUCCESS in 12m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/32dbf9668b9844898f61587d1b939fbe : SUCCESS in 15m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/df7abf7d366745ca80dea4574d292192 : SUCCESS in 7m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b8d61ae72383457197405d1181d5f19d : SUCCESS in 35m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f7a32e4d7fbc4e86a721206fe31c39c4 : SUCCESS in 13m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56f222d63ee74d37b84f1a75cf55f2e9 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5be772cc28dd494aa5a0b07d56e02db0 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ad5b3e55565c4739ba27231810fa0fe0 : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30aa2bbd1a624225b276d603a1a5caee : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21be426e74354286a461d46eae1cab2a : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2093781f8c964d3880eb7892ecb7d45d : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ec6bb60ecfe407ca0626ca73ab61074 : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"14d383a446bc5a711f1df7e676248cfe83c11cbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:45:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e7d221f86371155117a769566297789c71b0a435","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb73f141d61644f39343f08b04f153bb : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8dc586bf4ecb7f7618004c217cccf0a17e8e711","revisions":{"d8dc586bf4ecb7f7618004c217cccf0a17e8e711":{"kind":"REWORK","_number":1,"created":"2021-03-11 20:11:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/780102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/780102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/780102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/780102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/780102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/780102/1"}}},"commit":{"parents":[{"commit":"6f08defd003c3b48ecb6371b5e33a7c78f74a091","subject":"Merge \"update constraint for python-monascaclient to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f08defd003c3b48ecb6371b5e33a7c78f74a091"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 20:11:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 20:11:14.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 4.2.0","message":"update constraint for python-zunclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: I3fbc3d085c4c4802ae3452307755b584ee2cc497\nmeta: release:Code-Review+1: hongbin \u003chongbin034@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If6665443854e72e02a3da91eeb8815a117fc1a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8dc586bf4ecb7f7618004c217cccf0a17e8e711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8dc586bf4ecb7f7618004c217cccf0a17e8e711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
