)]}'
{"id":"openstack%2Frequirements~780106","triplet_id":"openstack%2Frequirements~master~I5e9b5c9042317eeb71c9b7172e1efa7a8e5b416e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e9b5c9042317eeb71c9b7172e1efa7a8e5b416e","subject":"update constraint for oslo.limit to new release 1.3.0","status":"MERGED","created":"2021-03-11 20:42:33.000000000","updated":"2021-03-12 03:36:34.000000000","submitted":"2021-03-12 03:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780106-new-release","meta_rev_id":"17c0b5df6b7789887fdc957f2a324c66ab6601ee","_number":780106,"virtual_id_number":780106,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 03:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-11 21:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-11 21:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-11 21:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 21:03:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 21:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 21:25:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-11 22:26:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f293a8eb98259acf0ff71f79cff9bb2ea3f0e9cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-11 20:42:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72b36ad22363e424a8ed680a8597024bc445dce8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 21:03:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a541a426d63e2f3bed4944cb266e116a67375c27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:19:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/60a1148d47d34c34ae5d75dd309a4ca3 : FAILURE in 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e22c442d480a8773b0a673c8077b90a391efd68d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 21:25:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fc59cc21c53b924a58b52c5ede610f1e00a71c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cdaed464d741d2bb1655319038505e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882f411c3ee9418ca947be5cef7cfc4e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1ab3ff2396f4030a2e7826347185cd5 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b437c60fdb2a497999c7c23115c57768 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87cb78205ea94ea0b1ffadb6c03b3b26 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e089689563ec4809a4a9dfb91ccb2cab : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23c754d5865e443897fedcd62039cb9e : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2135743fef094953af81331c1ec287ba : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/280fa87556164663a3e72679360cfd6c : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6fcc5e650c7640fe9e9d4f013e55965f : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1d94604a33ab4b45a5963263c558d7b9 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9e98fa1a049486988561ec16ada7b00 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70c5a582be874044b234b0cb038830e8 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a54a0595cfc462eb5966972d33c8796 : SUCCESS in 12m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d9e6714b3ed46c2b680068ebfde4f7c : SUCCESS in 11m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/02c8a29251d340baa0e957df5c4262b4 : SUCCESS in 6m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fda474409f15434ebb30a2a05ed0c27e : SUCCESS in 32m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3b5ac9959852470796b18ca82f484125 : SUCCESS in 13m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7bbcc0fd375409eb7e81d5cb43a9e6f : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11df16eaf2844e589270621fa3c4b4e8 : SUCCESS in 23m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9782bad505e04ab18c07f9941f61c5dc : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/344627e6b0ff48b28745ec5fb73161f2 : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/915b76a4fbb8443a91278335a31a4423 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/842ae129f6de47a39de3da5b2bacc5c8 : SUCCESS in 1h 02m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d2ebf99ff174d5b871710b14ea2e859 : FAILURE in 1h 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"420c59f3af1b23785dd44cb3101d3300fe730a86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-11 22:41:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56a9b00048da25a9362a7d510d9a6c7658a7d5fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:25:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/22385547b0ac448ba148734ef0ec2def : FAILURE in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"324d2d271497ab0f1e3ade2a94d6f515f20212ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ad00ddb6964eb5bf5d6c22db8a0fc9 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6393b566f2eb4e0ab69e6205e6f0ca92 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4ca4c653c64d40b4fe0d0d15092370 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53ba85b542e545b7b3a4872dc9666edd : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f83908f39a4b1083c9b4d0aadf7408 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2a4e09e32f04de8858c498cb8f06b90 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfa8afcd2e5548c89b1dbe1f6ca74481 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d42000b08adc4f909c2e67afd2c8f313 : SUCCESS in 14m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a132617375f049b5b27b3229b3b0e354 : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fcb6e369655c452a983902119ed4f2ce : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/42460e96146a4c9f965c840246124767 : SUCCESS in 8m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/319a7c1e5d8a433f9ccaf765651c51f3 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cf385124fb7403d8a0fb8088faecb3e : SUCCESS in 13m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f30718c1c2394c26989afc0fa4072ddd : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ff5f161fbdc54737923bc2b85ea575f9 : SUCCESS in 18m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/98b0c8a21f2f44df94a4d468a846deaf : SUCCESS in 6m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2f4f525064f4b679372eb9950f87178 : SUCCESS in 40m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/171949a3d8df4531a3e098be34e062fb : SUCCESS in 16m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e41739bb5c2a4bf09ccb8d52b16fa20b : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfa1019928564ac0a591f73a86e2ee18 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01a2c9deed1c40e2aa89fe5e53f38093 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7ade6a617eb43628d63893ebffe76c0 : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a11250496c81440d8d3db7af5f6a44ee : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/222dcf294a65492a977053731f709d44 : SUCCESS in 46m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8a1192d9375414d9e1b46cc7a83a57c : SUCCESS in 1h 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3f42466760c2ebbb4365f85011f5a0893cbc99b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:23:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbb24da04a854205fe42aba7d545e2204e7dae19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 03:34:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7632d0735e491792b163da52362c67 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa00acd1b84e41b88c01892d4dfca466 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b94488586042b49422eed2fbc1f36c : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b0b5fb456564968b42e11b3ada65d2c : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c594c97fe304bd5bd656a6bad2ed626 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acc001519c024101ad16cb1c2cf4d387 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca27a57323144799a5b6f9026c5deae1 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f92d564611ee4f9089cadb854f6ab82f : SUCCESS in 9m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2284af2556741ee93fce6850acc91e3 : SUCCESS in 12m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e199410010d547998b1b9c592a3abcfa : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/742aa1b0bc0a4b8eaa603ed956031a0f : SUCCESS in 7m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/663baa2e2c6849a794b80a04b1f9150a : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e36dc151824477891429a8be0138ae6 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53f122fdd7484ebc86424ec1c63dd7cc : SUCCESS in 14m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/07f6643f7cef4588b37f7f0fbc245b15 : SUCCESS in 4m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b9b5291977154aa4b17804b88e757b70 : SUCCESS in 30m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e7e9ce0768c34a1c97445a9deaa48313 : SUCCESS in 13m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e5e9807dbe849129f5750b0edb4e194 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ab90d65ac6a42cf9f83d9deecb6a17d : SUCCESS in 19m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01f8de89b9bb4f2da29aa972e39b6d40 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04d3c30b0cd84d459745fb14c7cc285f : SUCCESS in 5m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66b68382845e4d19b8c124b8b53a2436 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9097ef19b30345a1bfecb87a60a1374e : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/631421efc1ce4c948e3aed7dcc7edb40 : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"911fbeef1f2e272c478c59e04fda989a72e850b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 03:34:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17c0b5df6b7789887fdc957f2a324c66ab6601ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 03:36:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c9226954f742f880e3f8d38b328274 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12b25d366d1a9f01f08c10357f0a95ff305c3f7a","revisions":{"12b25d366d1a9f01f08c10357f0a95ff305c3f7a":{"kind":"REWORK","_number":1,"created":"2021-03-11 20:42:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780106/1"}}},"commit":{"parents":[{"commit":"6f08defd003c3b48ecb6371b5e33a7c78f74a091","subject":"Merge \"update constraint for python-monascaclient to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f08defd003c3b48ecb6371b5e33a7c78f74a091"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 20:42:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-11 20:42:30.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 1.3.0","message":"update constraint for oslo.limit to new release 1.3.0\n\nmeta: version: 1.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie0509f52d3f3b3872a4fa20b852d7e427c7c4d74\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5e9b5c9042317eeb71c9b7172e1efa7a8e5b416e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12b25d366d1a9f01f08c10357f0a95ff305c3f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12b25d366d1a9f01f08c10357f0a95ff305c3f7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
