)]}'
{"id":"openstack%2Frequirements~780141","triplet_id":"openstack%2Frequirements~master~I18584e102b548969078b78ad669d8b665363e97a","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I18584e102b548969078b78ad669d8b665363e97a","subject":"update constraint for wsme to new release 0.10.1","status":"MERGED","created":"2021-03-12 03:16:37.000000000","updated":"2021-03-12 06:00:50.000000000","submitted":"2021-03-12 05:59:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780141","meta_rev_id":"54c154cf57e3f512629b632983ddb8b012bf42e1","_number":780141,"virtual_id_number":780141,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 05:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 03:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 03:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 03:36:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-12 03:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 04:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd15a93d6ebece4f9f265ceac238dad2e7442a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-12 03:16:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1da39fd63e1845fd77e9e5b19805758fcd455d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 03:36:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7662fe72a9bae48157d06f53700038715df4747d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 03:48:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cc334ab36f23453fa3fd7b6a1617bc43 : FAILURE in 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"95f2341bd34c3907614ca5e97927f570480d784f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 04:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dc66b238b045ef8384e382ad3b2c23 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e32b5df8bb463298c9adcfb53fc162 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/920047a686cd4c7ebcd2ab345127ae6a : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b0fe9c95cf4c7ba8fc84a6f0b3e462 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee418f38aa204dd3b535850135821f4d : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e50ecf5dd9f846919fddea06fa6ce1ab : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b646d3ae080444ccbe07bc42b188c251 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c66a57e2ba0145329a5cb941ce27632c : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a4958f499601468aaf1132c620e025f7 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a5f4c55921543e7bcd9ed322711f783 : SUCCESS in 12m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2fe6925316c8460da409ea3c6767ad60 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c799694d8dd4ebc91ca1a9c35b78c28 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c014a61e6f944201adf002a2c8c065d5 : SUCCESS in 7m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64f5a249845e426d875a94435eb41b8d : SUCCESS in 10m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd3faa08720a438494e0f29fd7d42158 : SUCCESS in 22m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/23ecd2bca2004ad3b46a5411a326141b : SUCCESS in 4m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66364049446a4fffa0af5ae8334a0593 : SUCCESS in 37m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81b6c14cb1f24616837e30760074b2fb : SUCCESS in 15m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/349c9d9737964a7db96968e0075b2d57 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7091266cb0754913b2028fa9876a9a03 : SUCCESS in 19m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f7125c05be1a4275a1ab0f420b9bac0e : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5ae92531fd44844be15c12b1aac0059 : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/08bcb5fe55e54fe39fc56078b04b1d50 : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54b6b21660714fb29f105c9ffd33f1ea : SUCCESS in 55m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cfac5ddc81c47ddb049abd690ce0362 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e2611285a4ca09ae932fe79dd8ae0451ee54fda0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 04:36:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22f95460bc27c9e9d1c33d008cc97aea09c256f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 05:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1252950913d9486cb63b7048e81e72a0 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb90438b91f947c3a1403cbead0350fc : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a069e0b97c57415aa122602ada0dc5ea : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5acf1b5c52a748088da9ee67b227d01f : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65255589686c46cf8238d02485981e81 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04b72915f6f04c68af77387d197914e4 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ddec5dedd6b2453ebffcde128a07cdda : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58622d4d5fdb44b7a0b7ea495a9240bb : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/195a347327064542bde9ad5f96e86a0d : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1b43568651e48308ad1cb7e7fb5be56 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f30e27868564d289af37993a89a71f7 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fa864e91e274dfa852433040c53ff6e : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/902c475f78be401caa15f4746eb0f845 : SUCCESS in 9m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b636eda8ee65466aacc58b7db29b6c61 : SUCCESS in 13m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/23b297654c884fd3bccd1ccbe8a17416 : SUCCESS in 6m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cecb5303590b47799b0ce9596cab6869 : SUCCESS in 34m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b35b262919fd443ab081c7fa993a165c : SUCCESS in 12m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25c30e2be7bc48c38a191bce0fcea2e6 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/995ea2dd73b349478ef8e3a72d23804d : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a413f52631774d8fa8cc217da747bfee : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53ab5b05c0364c60bb30b721b7ab8e3f : SUCCESS in 6m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c9bc007e63dd4640817405d73e3d0b07 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a9ce06922e94220bfaa56a39e09bd8a : SUCCESS in 50m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7856ca1555f3472eb20ffa50795fd41b : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9c5d179a2b97df2b8a82dd84e8183c79fc3eee36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 05:59:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54c154cf57e3f512629b632983ddb8b012bf42e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 06:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f5941685da48acada210afd9e147fb : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0abb8e0ef61b969c7930ef0fe0833aa2081ade0","revisions":{"f0abb8e0ef61b969c7930ef0fe0833aa2081ade0":{"kind":"REWORK","_number":1,"created":"2021-03-12 03:16:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/41/780141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/780141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/780141/1"}}},"commit":{"parents":[{"commit":"6a79bfebbc6b424068324960ead8f8145391a7e4","subject":"Merge \"update constraint for oslo.privsep to new release 2.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a79bfebbc6b424068324960ead8f8145391a7e4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-12 03:14:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-12 03:16:36.000000000","tz":0},"subject":"update constraint for wsme to new release 0.10.1","message":"update constraint for wsme to new release 0.10.1\n\nReleased wsme to fix a python3 deprecation of getargspec.\n\nChange-Id: I18584e102b548969078b78ad669d8b665363e97a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0abb8e0ef61b969c7930ef0fe0833aa2081ade0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0abb8e0ef61b969c7930ef0fe0833aa2081ade0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
