)]}'
{"id":"openstack%2Frequirements~780227","triplet_id":"openstack%2Frequirements~master~I28ffba4fc6eed23e125bd43a6d897c77acf125a6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I28ffba4fc6eed23e125bd43a6d897c77acf125a6","subject":"update constraint for python-zaqarclient to new release 2.1.0","status":"MERGED","created":"2021-03-12 11:20:50.000000000","updated":"2021-03-13 09:21:27.000000000","submitted":"2021-03-13 09:19:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780227-new-release","meta_rev_id":"6ae8c39dcbb4343d1a93c31e1e460758c808b9fe","_number":780227,"virtual_id_number":780227,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 09:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 16:13:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 11:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:13:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 11:47:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-12 11:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 13:01:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 16:00:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-03-12 16:13:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"deee17458771574c36d298db75af5edafd1b9f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 11:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59cae7fb624c1e90669e76ff083f1567c8ccaa43","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 11:47:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"182cf9308ce5576dcf7a5af22619d3dd34a0060f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 11:54:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a6b364a4da19485b8eff6a8caab9be62 : FAILURE in 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ea2d7f69075ae4c9b45632f5422d70e879af42f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 13:01:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f28ad8c76b40f09d8f0018ca584cb1 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2897ab19f9465fb15420ed4041ede1 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b3810490a145568d7825a04c53c72c : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2f16beceb4464fbfd5850dd7a59f17 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba9b6a019324cc3b8cf8acca0f8c876 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67f646bfa645490bb13de5cb3993981a : SUCCESS in 6m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57ec302dd8c444c0858c4b7a4ed4038f : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa1e426a381e4a92a8cdb428b9908d9a : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50a8dffa3d164dbca768407367cd4ab9 : SUCCESS in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0cb13472a0144fa0930089fd322cad41 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3a552c0717240918ff19edbbdc5d3a4 : SUCCESS in 7m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/00e5a171c0cd4697a925caac1819f32e : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc24164591954aaaa6393641ae433ca3 : SUCCESS in 11m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8fff1a36d8f548ae8ddfb7fdd8f8b217 : SUCCESS in 11m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/45468c66284349d3933a020b435b346e : SUCCESS in 17m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/6801276c3bda488eb0e3f80382ba2649 : SUCCESS in 5m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb162990491744e5aaf36b6a9550475a : TIMED_OUT in 41m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/72b472fa57ce4fdf908157cd8b977137 : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4855d1329f1a40e780012c17c10158ef : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1bb1bee13014459ba3d5c277e2ee78f9 : SUCCESS in 19m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d22386b6e920449a9c1cb9e411431797 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9d36acc6fa24a2bb6b91cb42682867e : SUCCESS in 7m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07880378a11a4b0a8c5692462d863561 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57457492ff4041cf94e2bcef4006cd0f : SUCCESS in 51m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/287c13a5fc3c44b084d3a5d9b0633fee : SUCCESS in 1h 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"297853fa4e93ca2ab2b0be1f55997e3cfb39bb49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:00:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4969fa6d90c6d1fe6f011e128e705ed3a70cf16a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:13:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7c1e0958580f04e8756f194daf487b06f5df99e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:43:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6b2a8fa9cedd4c8086ff7041c727cc91 : FAILURE in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b1e72574716707feab83e87fc76b83eaa79dd8e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 17:39:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f09d26234f9461ea682c3b53eb758da : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac621c3b6b44a16bd56896852fd3884 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f6804a93e564b44b3aaccf7fa52c002 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd4f4d448af495e88cd3130f449c574 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92faa605e545497f816ba97487a90529 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/857fd3cd210842b386ed72a3a2449f51 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13f58323293142b2adefc9c3ce441b54 : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f354ce2234c4c80934360ca198ef55e : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f027729fd82d47ee9ba9e311d3554399 : SUCCESS in 10m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/693e79c1ac864a8bb36eb87acedafef6 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9cfee95eb2f14d02a5aac12d70059c49 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f163b66aac614e579bb4c7fdbf142fa7 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/878bed89625b4a3ca4f10c698297739f : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/94af00e0f6624569a42d5ab55aa6f403 : SUCCESS in 11m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9b0182bd27774b5dbf1b3ed1089e67db : SUCCESS in 14m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9fdcdd6f160a4490a73e65de477d6745 : SUCCESS in 6m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4a2456e5271f4b82b580f2f5464221b1 : SUCCESS in 34m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac2eb01e36db4dc3bbe51c46b51ffe57 : SUCCESS in 13m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea5db910ed6145ed9753da1d89c85de4 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12be32af28c34a01ab7ea641cf3aec4c : SUCCESS in 21m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/006e1cc8e46d4c06a3a01503db3d549a : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6312132210ab414fb9365ddd8cf77120 : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c252b48ba0d43d689a91c07baf01f80 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91944af2f2fe49489acfc5a9811f7ce3 : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba88732cef0c4b6e879b958aaa705447 : SUCCESS in 1h 33m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f9a86a529d44cd882d6012dc2c5bdce6778cf4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:26:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd5bc0d8d276428cc4f3a6ed8abbaaa100869684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 21:12:38.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f44d0ef48eb73754670f278e235d50f6fb967ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 23:38:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ec23edd7fc455e8bb12c7e00cbf337 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f7d1228ba6403c9ff807caf284596d : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d40bf82a49d417b80868574f2ef7628 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d10df48a0794f44a3faf0a85ca9a353 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b2c09bac1f34c2fae3f59fad2e4865f : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf9332dd9175432bb4853239756bdff5 : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d9905b9ba94488181b953e64c1eb330 : SUCCESS in 12m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e64547efb184bae861598d90c112669 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f0d448650954c3dbd9ac9d32462fa6e : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f33e1ec44b194e6f8b0b0fb8af1fd0c6 : SUCCESS in 7m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8beeb50ce3e24a4b903a888fe7bb19e0 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6edf95dd13864d0bb84a8653354e2282 : SUCCESS in 11m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a35f50eb8f5f4080bbe400748eebd1f2 : SUCCESS in 15m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f80207501b3b4988b76ce349bde5e510 : SUCCESS in 17m 07s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c51f7232e5144b0398349358ff761b6a : SUCCESS in 7m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/95e81c9845384c02ac9272970b097d3f : TIMED_OUT in 41m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7ce85c4ea0a497c808c6607490017e7 : SUCCESS in 15m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63991d2e91df49599998d93fa98c6faa : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27e9beb88597409caf313d14bdb91c49 : SUCCESS in 25m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae15b1ee21f34484b6b9a562d8ba8e0b : SUCCESS in 8m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b8af06001a0143eca4362ad158b0c731 : SUCCESS in 9m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/62dad47b734847f883549a316493bcff : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71710d20b51e4dc7aca3fe0d57b17484 : SUCCESS in 1h 07m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83971e563e64dedab4afb80afff27fd : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b88bcea8e804b826507079d861edddda9f3bd847","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 00:04:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8c6ab818f86bdedd084c5d8ff594ec45c9ce93b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 00:42:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ae39b02cf3fb4881968203e18c1c480a : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d84ccb7f9d55d439ed4d999c640a0b487997dc6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:29:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d9d8b6331b4c78956d18d5bf0112bb : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dbe020113e3481da93118707d94bc65 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9439ff94d2ba4967b174c05ed51ebd52 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0221060a2c814036b213916625f328b8 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe8a620bc0d94a1bb22e5828f6c397e8 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/accf32629b6b44c7a0da84438c91a624 : SUCCESS in 8m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d37e150ac1644834a62f39630cda0d91 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c2165867b9042af9edb66a2de3aab0e : SUCCESS in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d24aa00b87994943b3637c8bcf56e7b8 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/28ad537bbd7d4aaf966d29b2d08e0f8b : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6fc9bd49ef84ae7800c6fc8714dbcc3 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8322e8b4590417db5d343c43b16987c : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/745c3508b5ea4e199b7b500d50789da8 : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b7cb01c522e04a03a287520aebf3d357 : SUCCESS in 11m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c09928f870fd47f5a6caeafa76f11203 : SUCCESS in 11m 48s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/07cbf56a2c2c4178ad81aa88275eea86 : SUCCESS in 8m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a04161ef76749a6a586104ec4a128ec : SUCCESS in 33m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/487a0225e32e4c6da4242b55ccc69325 : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3eb37c0b49d4976bb609fe208e79d55 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb56c26004194951bd3c4453475b00cc : SUCCESS in 21m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/88ea4908179e4e5cb8ecb2f068b5d17d : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00cfd7bf22b446a89d3fb67913874e58 : SUCCESS in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f7fb42c4c18a446198cd3a28cc69a660 : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78634428201a447f939f75c23a475554 : SUCCESS in 53m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a3600adc49041ff94b907774c1cf1ab : POST_FAILURE in 1h 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ac115ff0b7864ec371aeef794a4e220967d04b72","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 01:36:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5fdc9edc02b9f1befa0a7a02167f94df204b595","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 02:09:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/30db713ccf3448b18707222ee1c626c2 : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cab86c7ef27a837bf15c803bc4e32965b55b9c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 02:56:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b02721c71bb4295bcb41c02e0d150d2 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a2a2eee30947c08300a082cbd0da9e : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3ac764d61084ca5b8e41131c9a13b9f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c7de61341a45c993c3c8185a5c0b17 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8099e05f57a8447c9b0ad55fd4823212 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/360a72b1e2e94bb4825066477a6bb272 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c32136b86927448c8346ccc1095ec641 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a34ce0482d84a4c906095b5362fc651 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab4b17aba30a49ad8c13e54609915593 : SUCCESS in 8m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa62e09432df45f49972ea6473ff56c6 : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a60babec6e5b43d8a44079ef258bafbd : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfb3b62519394a48ae02a708189477fe : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1376dbf7dc442e1896efe9c4c63ea63 : SUCCESS in 9m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6bac7fe7bda1452d824053dfba60b39f : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8184f9fcb5764f2b8446d92459329f6a : SUCCESS in 13m 31s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0138d150c79c4f4682893193d1ecab11 : SUCCESS in 5m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c0ae3e3b5c9d4bf2b8b3dffca54949fb : SUCCESS in 30m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/38797e7cab104f4cbc4fe3d96dea4505 : SUCCESS in 9m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52eca1edd0d3474f80da89b13075cc9d : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efec910a19b2417bb1c0c44818aafe82 : SUCCESS in 15m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4fe7d01447e44bb18b407d05ff1b3824 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe9d8e047b364c64bce14f479d56ed88 : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/070b6f857ba448f291329d58fc3eff4d : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c8c11d3da9441d78b7a60e6ab9120f4 : SUCCESS in 1h 04m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fe9abc5dae447c5b621caa850dde596 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d37818b144c29c1dd2540a55ddfb25d8577fb864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 02:56:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13257555ea36e562026eeb73359c16fce1c54464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:19:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789971ada2cd423d9855acbceed71c61 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce146b5344a84f41a5b3f6e7d30b5cbe : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92833b4d22ea412db72d2bbe3f832c89 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e82f1b85dedf4049b67a5246f0b53a26 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e15b14c680984c909be3af182ac2813c : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/594693bcf14f4dbbb3a674323efbdd4a : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7f891396c3544cab5091ab41b6093e8 : SUCCESS in 9m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e332bb15b6ea4c11be7403cc1625e8ad : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c251f35747df4b4ebd0375bce96dc31f : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/02f0767a297640a390c7a1556b99b5d9 : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2816df682014b428583436fb7b755a2 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63dea6e600914037a3447c7d9d886950 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/76dd64fea364469daa9c10421ba524be : SUCCESS in 10m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a8c69e83461d422b98c07f5040dc2c1d : SUCCESS in 19m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3a3057aad9fe4e93a77306699a676ddf : SUCCESS in 7m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/744f66342a594cb7a40e1f628f8758c3 : SUCCESS in 31m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/262197f5bdb3422b8995e2bf529083ce : SUCCESS in 18m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9738f88671c44c9b92dcd61b479310d6 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6857172891a44b109878b1abf4639a56 : SUCCESS in 22m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/59578389d6de4762bf3cff8c5b8b731b : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3666a76d6a514fe98bdf0234665c4613 : SUCCESS in 7m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c4368845aeb488ba2804698a73f754b : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/572e615bc317473ca9741a1b964785e2 : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac4528c7ded9466e886a63afc9dd8659 : SUCCESS in 1h 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f70bc9405561644a8a2e4be9db75ca1e51e203d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:19:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ae8c39dcbb4343d1a93c31e1e460758c808b9fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:21:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68acb19c156c4a0d965aa7fb91d8a0d9 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc464ceec218546ae568bdf0078e536e58921e21","revisions":{"fc464ceec218546ae568bdf0078e536e58921e21":{"kind":"REWORK","_number":1,"created":"2021-03-12 11:20:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/780227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/780227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/780227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/780227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/780227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/780227/1"}}},"commit":{"parents":[{"commit":"19fba96498eeb97e3030cc0838597c2c2c38f133","subject":"Merge \"update constraint for wsme to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fba96498eeb97e3030cc0838597c2c2c38f133"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 11:20:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 11:20:46.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.1.0","message":"update constraint for python-zaqarclient to new release 2.1.0\n\nmeta: version: 2.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: I23af9ad605ba8c286624b769a840a38fcd133637\nmeta: release:Code-Review+1: 王昊 \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I28ffba4fc6eed23e125bd43a6d897c77acf125a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc464ceec218546ae568bdf0078e536e58921e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc464ceec218546ae568bdf0078e536e58921e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
