)]}'
{"id":"openstack%2Frequirements~780252","triplet_id":"openstack%2Frequirements~master~I94a42ee6c1cde7ac42952b306397657d9d300a8c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I94a42ee6c1cde7ac42952b306397657d9d300a8c","subject":"update constraint for ceilometermiddleware to new release 2.2.0","status":"MERGED","created":"2021-03-12 12:46:48.000000000","updated":"2021-03-12 18:47:52.000000000","submitted":"2021-03-12 18:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780252-new-release","meta_rev_id":"8666020d488194265ac1c3f45f86c5f05622e627","_number":780252,"virtual_id_number":780252,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 18:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 16:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 12:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:00:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 12:51:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-12 13:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 14:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 16:00:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"103dba823dc7d9f0dc8dee4b7106df63a1c1ff13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 12:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"021c542e08867729bcac6ec1957b679c2c4afadc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 12:51:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0dd82702cac0d1c5da01ab745df4c9d536a5afb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 13:28:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b28720d3ee94902989de5109a2d5b8e : FAILURE in 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cf5d7036dc030c368243dcd1ff7853cecd85bc86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:11:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56091deca74c4cb9b3d4e097f72104f0 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5d0eaeba37430d988a4d3644d6c71c : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b69936823b64ab780cffa9eadbaad79 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab3776edee94d449d08b9015ceb0e2c : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/276701f70b33473ebf4ef0f38872e014 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8df6a479221a4f2f8b2a2034adfc99b5 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e47bfcb868b46999347b84d073520da : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c8fc1c537624567abe319ddf9a81bbc : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5942cd9694414952816168a449bda351 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/09c5f7b6706f4136b19de970a47745ad : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe122919d35949cf8ca574be6540d105 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eb8c7fef0a074ddfac9df9930d1c54e8 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4733ffed325849c3bb17ab93ea961659 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3a8c9a8d8a64657b8f33cd554b55762 : SUCCESS in 10m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c40522745da942f0b828f2958f3ef087 : SUCCESS in 15m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/679e631505ae4187a43f6aa22a2d6303 : SUCCESS in 5m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eea80498db0c44d89d13b6a5bc620321 : SUCCESS in 31m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/028910ba7e64442f9e779bf5fa834656 : SUCCESS in 14m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dd1582277fb4e09969ce4f569d534ec : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e407a7991bb7438f8f3f01b2c0716881 : SUCCESS in 20m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f72dfd7372cb42c58fa2478f7cac9d00 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6467e06d5a6e4d77bb15ca76061e3fd5 : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/087176c616cd45fc86e28e951f96494c : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73fc92eddf69476d8d253a3fbb1e080a : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44e70d56812b40079e6385ccb8af1b6f : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7ac91fad658304f03b724d7141a17e6445220c33","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:00:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"085626fe3478860bf5bc97fa7168a83ecef88cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:01:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"724a14ca51785230f4262c9c29c567eed9c98d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:45:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b321c39b5ffe4f64be3089ca161bbf9d : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb44230a30c14ceeb812c22709759fa7 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/487f8691bf7740fda30c6f31a413ce04 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb33f22c58d4d8cb9f5b66fa7b810fa : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb9e3a80004b436cbd69ce5cd68450bc : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f47eb290fbc40b696c46be1f4717f53 : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b0d87265d95748ba80f9adb256a32ced : SUCCESS in 8m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50f8e5ff7e604a78927690299466c126 : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8078d32d12804e41b4d721d6e67f0f9b : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bbf764525c4b4df096ca4c4a29b99fa3 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a53aaa599434a60a607f09cf868dc2b : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12cdd17a57e1431fbd475b330a6feb92 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7caf16bf7aaa4711bcefc8ca66faa53f : SUCCESS in 11m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f180ebcac51e45929f786baee270dd21 : SUCCESS in 17m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3c39cc018c0f48519f1a882f6beeb0f1 : SUCCESS in 6m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/27c1ac1ffd2b4fe38d7ddbdcb5946c4f : SUCCESS in 37m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b05a694d5fbb4c54ba16a24b3dfb144f : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9d6e6d5945e4bb38248ab73e8f652a5 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a1457ab264d42b588788573fb061366 : SUCCESS in 25m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4435e23442674e9b8c2b522631d9a143 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/815cfd03051843a79c0c75c4fe8641c3 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ececdf7ac06740b1b3e6bd51db72d78a : SUCCESS in 17m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d691731d17f421985ca60abbce32fde : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffd2348363894da39c051673df1b9587 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d5be4b729df80ff957a3e0ea5b3880d4c58e27ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:45:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8666020d488194265ac1c3f45f86c5f05622e627","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:47:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e55ffa427124c8fab1e36c29eef0b01 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a8bbc09b915dc87e22da4af62414a0ca35e62e6","revisions":{"5a8bbc09b915dc87e22da4af62414a0ca35e62e6":{"kind":"REWORK","_number":1,"created":"2021-03-12 12:46:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/780252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/780252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/780252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/780252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/780252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/780252/1"}}},"commit":{"parents":[{"commit":"19fba96498eeb97e3030cc0838597c2c2c38f133","subject":"Merge \"update constraint for wsme to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fba96498eeb97e3030cc0838597c2c2c38f133"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 12:46:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 12:46:47.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 2.2.0","message":"update constraint for ceilometermiddleware to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: Ib0a1fd078ed84eb88ca42602f6c2c0880cf64f6e\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I94a42ee6c1cde7ac42952b306397657d9d300a8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a8bbc09b915dc87e22da4af62414a0ca35e62e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a8bbc09b915dc87e22da4af62414a0ca35e62e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
