)]}'
{"id":"openstack%2Frequirements~780253","triplet_id":"openstack%2Frequirements~master~Iff3a44c0e2a55be950bc47d32937ec4355edd39e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iff3a44c0e2a55be950bc47d32937ec4355edd39e","subject":"update constraint for aodhclient to new release 2.2.0","status":"MERGED","created":"2021-03-12 12:52:37.000000000","updated":"2021-03-13 01:32:32.000000000","submitted":"2021-03-13 01:31:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780253-new-release","meta_rev_id":"66333f5e0b89295652bd4e2310c0d6c5fa904540","_number":780253,"virtual_id_number":780253,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 01:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 16:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 12:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 12:53:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-12 13:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 14:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 16:00:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c2eef4f6d5fd6e5b7eed1c268211182052414279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 12:52:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9fecb995dde743b080ea92302237176db7d6e5a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 12:53:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73509f985eb64ad5504f0686b8267edcff2b09ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 13:29:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de80dfe8257b4b35975dca96ef990859 : FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"25642ae30eb3400d0c1d425f01fa3e0e35044a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:11:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7927621af1c4b198422763deda30d34 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8d50d8d9bc4328a862dd5e9fd69efd : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2dfc1538c344707a5198a4ba5eb8c0d : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d4965e329942f4b767a57e06532a6e : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b089a3cd2a5343cfbb0d789450cb758a : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e6527ce64df440bad640b4403c9414e : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f274071bee864ecaa8376a22448e389e : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/922f48fba4d541fa948fde36b159e734 : SUCCESS in 12m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c424ec27910e4113a051a7d079ec22f7 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ab7dc41e14784796aced08c734e1c301 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d62466012c3462fa2f69a3c2d5c3c3e : SUCCESS in 9m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/06f3faf548734fd99219545d5cc65651 : SUCCESS in 7m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b824fe5a73a4a659359dcb4712d4bf8 : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d72bfbd402d1412c972e51202d6a570c : SUCCESS in 9m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/97863995b9f649239fde6bac17bb533d : SUCCESS in 14m 23s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a3c01e88df404454a1815536563858d2 : SUCCESS in 6m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/475c3421380c4437b686565688395dae : SUCCESS in 24m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eeb5abfb74d240169421bd81a0c5d4b2 : SUCCESS in 20m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cb607290ed84cf183a9c16549682300 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11d0ff22fe0a44b38a7b8d2c08f23e44 : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4597fa0954a04caeb58befacc9a31425 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e45985d95084f65bd57b5bbaccd9a5e : SUCCESS in 7m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ffe5e4e8e7b54906a9af57aa9a5f2403 : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45abf24445ff4f11ae50dd4c9b5afe80 : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/939a21f0df734a6aa5f87b715bf9aa78 : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ff30e8e6f63416f760b46813a6fedbf56c7b7734","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:00:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3f1de88de69d37597bbbd8c249c80ab344438e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:04:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e5d79855c4da31877b3db7f411da6ce9548ba28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:46:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93a1a4c67214396a12bfc5b3ccfa839 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d565e9278d4a17a2ed83e75fe2d5fb : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba1251edad0444a857db2ab9e2ccb0c : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e2134b418f34909969831a54e72ecd8 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35554be70e1f480faaeb0741f4cf17ce : SUCCESS in 6m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d2febceadd146488859d5bde8b81406 : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3cfeb1d9a0ff4fb69ac0d33ec5da61f8 : SUCCESS in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/928cbec8ed614bc482615f5d7960993d : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4457652df00f416f808ceed6f150f18f : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ded0a5a76db64c6a8e42cabbba3373dd : SUCCESS in 7m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/47024b60d7044070a7bcf880bdfb7cdf : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/674018f897ee44e9bba8cb90c1a52400 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f788a65d96e447b9b6f2bc39da5d41df : SUCCESS in 10m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/57ecfc910d494ae8b6f15e68024c6eda : FAILURE in 12m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/baf7a67010ad467e902290422bb1a160 : SUCCESS in 6m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa56bd75747244f39b3e2da36b459bcb : SUCCESS in 33m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/385fdcef53b24a9e9e1738c516d08609 : SUCCESS in 15m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a742c0a17e045459f86628ea49a9267 : SUCCESS in 6m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4789f4f49f7442809b8a05edba471047 : SUCCESS in 20m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/885a1f38d2e34f49a320073903ea4ac2 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a1d848003984440991392f4aef0ef29 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e52426446d1e4c73aa9d5447b31ad5d7 : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e76e396a1494a069b03a9930cc58da8 : SUCCESS in 56m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afcb693af0cd4c679d9cc2e25725fc94 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ecba65d7373bd8eda266d8182c8589f9bde56e26","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 18:57:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50e3df1d34030574a456aa8c0b38b6a2af211cca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:35:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/111cfe0608c543a8bb99149828c9f556 : FAILURE in 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8ab55ab68a87b0f5a5fcdef06171d2509c358d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe396830a1e4ee48f0c63a689552f37 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b94edd2017e1403c892d468d42f23668 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1611266847d4d31866cc34e88c5f432 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3dd56e844d441489e1c7969554553e8 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45cf2d8241354cd8bbf3be488ddc61fe : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d492d57df8274f888cfc3b017fbceab3 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ec3cec059e14eaba55471e26b49d872 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef7e9ce84420477686698dab81de2e17 : SUCCESS in 10m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57dd4930e27b4771bd10c58f18fbdda2 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e5da1dd96dbd41d184ac0ed3bd956566 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62f24fd446aa4207853ba4e3fa1ac0a9 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/573711a8c37a41ca9ef292c73666b5d2 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6153362211f468f8ac391f54927a3af : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63b1a2742c414df8a3fc97767b408376 : SUCCESS in 10m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/378abce084ae4fe28a2076d0739dec36 : SUCCESS in 15m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0d6ae96175894bcf9592d94ff280c46f : SUCCESS in 8m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7f2bc0c38ca4cf499a62abe44d7ea11 : SUCCESS in 37m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4a976e216f1f4069a661f5c70c9425e5 : SUCCESS in 18m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a51bee9340b4f54993abefa6185eccd : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bfa03867d504659a9890367808dcf01 : SUCCESS in 24m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f7ec702597dd45ff9f2e93c3051670bd : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fda6891b05245bf8d087ab6284a865f : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b1a2561c578b4a508463d827d2240087 : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fbc2639b9694daca5e9a50e8d232a37 : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba2efa917289436291619dd4bdbeaa46 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b0db985bd41db80763ae6ac6ef26001ce3694ace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:50:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1322fecd10c8e5dcb16cf582e31f05fd53e78dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:30:06.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7d8fb782d00cd2b5374e992047d159a5f8f6d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:31:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fa468968a443e08d5e762399079a7f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58745d51fa52422982cffd5c675770d5 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a057799b504088875762cfa9df3d0d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa28b373139546bab7dce1d1b0d74009 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84bccb7026234c1ba69639c9f4fd8308 : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb65f13cad9f4eda968332a4dc429a21 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d0420cdeadf493abad132926b764e8f : SUCCESS in 10m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b16f7205a48b44da81df35f757eb90a6 : SUCCESS in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7b3550d228c9458da8907f68bba7e45f : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a720865c54dd4a6c9acdbb36b884f01d : SUCCESS in 9m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13f2a6b1354c43a4b02f4dd428e72f87 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/241ae8753cfa4237bc27be3df84c7dfb : SUCCESS in 11m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2c3a4fa542294899b71cf2e749c0a152 : SUCCESS in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/718b969a0c244de695d06c813dbc119d : SUCCESS in 14m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5d808c7d17164879b8f4adaf3805f598 : SUCCESS in 6m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7115514222a6424690441128021e9092 : SUCCESS in 30m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7736e681ab35472dbd46702e18ee98d3 : SUCCESS in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c25f71585124f84b76663eb8c2c6f35 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/512735f7ba2946c7b4504c0a1e15dda7 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a5c983ce8b954a8ca7a9cad7a1ee26e9 : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5118fb1b072c4380ba971ee2d0be6065 : SUCCESS in 7m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f53aea0a4de4684b37e39ad866c8fe5 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff0400d1d34c4969b99281ca4c0423db : SUCCESS in 42m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1778669021e948b8b4920440d48f34b5 : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"758d8fb3a80c9e3ccb022d5f9fcdc25608c06df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:31:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"66333f5e0b89295652bd4e2310c0d6c5fa904540","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:32:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b8895d931445a4816ca3b893e9ccea : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"260126f263b1f53b9743381e57ba726b7316f929","revisions":{"260126f263b1f53b9743381e57ba726b7316f929":{"kind":"REWORK","_number":1,"created":"2021-03-12 12:52:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/780253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/780253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/780253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/780253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/780253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/780253/1"}}},"commit":{"parents":[{"commit":"19fba96498eeb97e3030cc0838597c2c2c38f133","subject":"Merge \"update constraint for wsme to new release 0.10.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fba96498eeb97e3030cc0838597c2c2c38f133"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 12:52:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 12:52:34.000000000","tz":0},"subject":"update constraint for aodhclient to new release 2.2.0","message":"update constraint for aodhclient to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: Ib789b2b3d9bbbf1acb723ec21ca3c990fa8d6489\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iff3a44c0e2a55be950bc47d32937ec4355edd39e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/260126f263b1f53b9743381e57ba726b7316f929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/260126f263b1f53b9743381e57ba726b7316f929"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
