)]}'
{"id":"openstack%2Frequirements~780278","triplet_id":"openstack%2Frequirements~master~I96163945e7291f598a4d260854bbfc3e2b874ed0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I96163945e7291f598a4d260854bbfc3e2b874ed0","subject":"update constraint for python-masakariclient to new release 7.0.0","status":"MERGED","created":"2021-03-12 15:20:11.000000000","updated":"2021-03-12 22:43:38.000000000","submitted":"2021-03-12 22:38:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780278-new-release","meta_rev_id":"73047e865685328024f3339cff959e12629e006c","_number":780278,"virtual_id_number":780278,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 22:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 16:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-03-12 15:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 15:41:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-03-12 16:00:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-12 16:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 16:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"408dd8890a9f11349c47f9acec0bd8650938bda4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 15:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"128fe858b9b39261f958e2f2ebc032e53230c733","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-12 15:41:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21b1e926abeeca1e1ec9f5253bb6f108a770d3a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:00:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fd6f35eb8a00538728e4d5f42726a39745d0d52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:03:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5b2f5fca14ce44adb88ab1f3ec98432d : FAILURE in 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"acae57648122450eca1ac1b570e50d9fd75bfdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:47:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb4ab5576ea4a898297d4f2939dce44 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46248edbe5441258d7f6783794f13b8 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aabb1579af96406a923a4d21d4b257db : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575a94f0b103457abe145eb06b9e78c3 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e5624d4d8964be5915d39e27e5c8dbe : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b27f142627fb4229bc698c5231f208ae : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/470ba02b27924e809e20ea2b47abf31f : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39ce5a2486a44fe5a95071747c44b2f6 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37a83f2ed92346d88e0b064deb926ee8 : SUCCESS in 12m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc63f306ab5b4e0d887ab2932a87ba2b : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59f345a477f2411db9c2f7a11dbaa822 : SUCCESS in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a0e4ceaff5140daa6357c130912c59b : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84e750a92c4544e2a40a7cc4cb963cea : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d80a85b5a7584e8a8cff03a9f1f089b6 : SUCCESS in 13m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c525bce35fb44a77b4e42052faf14aa7 : SUCCESS in 14m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0df447b46ce44c528471e1caa7029fcf : SUCCESS in 7m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9d369f063ccb40a68f82141e92ea938e : SUCCESS in 36m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3a0e27c547f449259e0c90482e9a1dc4 : SUCCESS in 12m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b54e6734a17401f905bbd19f24ef464 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc3f09c9863040dc9198252981f0fb40 : SUCCESS in 21m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b8cc16c6dba94511a7f28eca347543bb : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b61333cf1fa74d1aa316ed873d965fcc : SUCCESS in 8m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a9e3e35cad3a446893a97b315c942531 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53aaea2421174b6ba6f664459c327b75 : SUCCESS in 55m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e19ce5663d024de98f5c0c144f9662d0 : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"45e427f6736482a31f32427d2c7c95ec1caae528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:24:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f29c9930c48da2ef86b1ca0b978fb7ed49361960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 21:11:41.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9359d1e6e92f35705e2a7456c8c71d2dc78f2fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:38:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6e25499a7149f8b6ea09a7f5bb9889 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca205ae03a84d3d97649714d9568d25 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e184ac1c6b4887b337fbe5bed12e5d : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b82006b8bfee458891a5ff1182eede50 : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e8a325b34fd4476b72a3da8b7557c1d : SUCCESS in 7m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1cfcb6fd1e264ef1a861a4488892591e : SUCCESS in 7m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c7d281d4f189491aa1bd6e13d7e9649e : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8d87c2eabe24b0b8f9572f45a1c0be1 : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82c113bced1d4428bbfc385b9a2de04a : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d61752f1037f4734b842b1b265319a28 : SUCCESS in 8m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b00d8fd4aa2b45c285992ae1d256d800 : SUCCESS in 15m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0efe0f030844bf48aa32d12be4140ef : SUCCESS in 11m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9b25c6d0733943e9abc09e791ca9c591 : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9c0d1ab7239e4da4bef0d99d4418d599 : SUCCESS in 17m 21s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2664f8bb4c864a9a850aa3d936d5cccf : SUCCESS in 7m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/00c16a5be42c4653bc48ab5c50094e53 : SUCCESS in 29m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d70393df80541d48072e9f4876612fa : SUCCESS in 14m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42d83b5e21c54d3b8343bf8b40bcf0fa : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95fe702e5cad445b9db39c23ce9d973a : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7bb2dfe794bc48de8608213f579ad83a : SUCCESS in 7m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/499ea5820b2f477d81f3f5a27dd9a969 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/91920067812c42ce89c8c1fa0a441244 : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc012fb4de843f8a4610dae1a1b955f : SUCCESS in 57m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a451a4d0f6114254a209b645c6d41f78 : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8d74162b2490b52f22706801c9f3f6025a49c301","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:38:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"73047e865685328024f3339cff959e12629e006c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:43:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da09ffe423e24f4e81e76d7c40290349 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da42231faef2c33c7d13d75d25c8ad680d4725dd","revisions":{"da42231faef2c33c7d13d75d25c8ad680d4725dd":{"kind":"REWORK","_number":1,"created":"2021-03-12 15:20:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/780278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/780278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/780278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/780278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/780278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/780278/1"}}},"commit":{"parents":[{"commit":"45e4013e794892d234ab004be7093644fa776cfa","subject":"update constraint for openstacksdk to new release 0.55.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e4013e794892d234ab004be7093644fa776cfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 15:20:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 15:20:07.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 7.0.0","message":"update constraint for python-masakariclient to new release 7.0.0\n\nmeta: version: 7.0.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Commit: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Change-Id: I28c244cf6d42010741819535d3cdf1dd9567f0d7\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I96163945e7291f598a4d260854bbfc3e2b874ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da42231faef2c33c7d13d75d25c8ad680d4725dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da42231faef2c33c7d13d75d25c8ad680d4725dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
