)]}'
{"id":"openstack%2Frequirements~780299","triplet_id":"openstack%2Frequirements~master~I46eda7b2cf81049a3790786f418797a27cf53d29","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I46eda7b2cf81049a3790786f418797a27cf53d29","subject":"update constraint for python-ironic-inspector-client to new release 4.5.0","status":"MERGED","created":"2021-03-12 16:53:48.000000000","updated":"2021-03-13 21:38:12.000000000","submitted":"2021-03-13 21:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780299-new-release","meta_rev_id":"f1f4dc1b749d5f5f5df9129dfeeeaa69c3a8ab73","_number":780299,"virtual_id_number":780299,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 21:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 16:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 16:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 16:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 16:58:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-12 16:58:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-12 19:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"713834bb55cb317b085070a4226953ff5532ce94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 16:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b101eaa2772849e294d472406d73a0e80a8d615e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 16:58:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed6d053e293b413b27686d41290870eb672f0be7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 16:58:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"379c8250d4ec1f7b811fc0cf50b40d4a7b895627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:22:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3be473d23742a7ab7b601274870adc : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345f8317d61943078a1abe0ce499fc7b : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb67e82ce89648ed9ba609e3206a233b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/501306670db14e82a1dceff0d2b1d8ba : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa647a93e6804e43be0b45a8be0ca14f : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/122e7554db09410b80c9c8f647e9dbc0 : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d7b36475aa04d0ea519e08f08cae463 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b440e672cbf4ed68d5a278a7df1bd83 : SUCCESS in 16m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0829c50ada92443eab78e6fe6202d607 : SUCCESS in 12m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc52be85f54745219a5c9c0185ef298b : SUCCESS in 17m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d99874b5b9a415da40f1a2a5c564335 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3410d74e7f4c49cf824264cad395be49 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0eccf423a26a42068347b9a5862b8689 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e3c4238b6245449e8a7bfe7cb12ff09a : SUCCESS in 12m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/50b231bcdf8249978bff5b4a5b8e036b : SUCCESS in 16m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/91428b7d000249fa88acb9e0cc6f785c : SUCCESS in 5m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2b3003fa75c4489b9945b1a0a42f539 : TIMED_OUT in 42m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6425f7b60f4e44c59cc49f8aeb6f2d0d : SUCCESS in 14m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a981d77d8c24115b512feec6aef2526 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce87ead5ab984bb38c54d2b9bdc6ca08 : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d57acc6cfa4c4cea9403108bb0503615 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a2d189afac345bdbc9956068f77e89f : SUCCESS in 6m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b4bf79a72dc488d9ea4b2dace319f1f : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef47a4a61bb94b6789a12e6a268c7db8 : SUCCESS in 45m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9eb9709959ee4844aab204c9908bbbf3 : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9b3e0cff9fbf4a9b1ffb1c76a4a12e564266ed66","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 20:43:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"17b60fa534393ff80e3cdab1557743012dbe0220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:58:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9472ca573d006981e638f45aa14003498ede4e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:54:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856dd7730e51443b9ba94b93210a73e6 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9ef0a1efec440885c17ef653c0a65f : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e9405da7d2458988a3f1c0bce76836 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515c054fc4944531a7ccbf2f85fb07cd : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/496f67b7a25f45058a38192c1b9d0d5d : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6a8bd8e11074a4088e057add6a39d0c : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9defd9cfb44c44e3a01a6f21d866806b : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b3b74a559aa4ef1833b51b4e3218650 : SUCCESS in 11m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d6efccb2e8f044b3811baa078860f6a7 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e36847f1316143d28f076030f9ca14af : FAILURE in 14m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b9079bcb2ad5482d8c72cb22e615a8a9 : SUCCESS in 8m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/415af7a4c85341c29faa89a7ed59a645 : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3243f347c18f4908944c92417ba0d8d8 : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/59a487e913d14c00b43cf042feebede8 : SUCCESS in 13m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/98d137ebbbae4a3a9492ecd33781b21f : SUCCESS in 15m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/25efb9e0e5fe448494ac419e9f87773c : SUCCESS in 6m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bd2c81e1c1ba4a11ae49fb698589d4c9 : SUCCESS in 37m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/457c442f37a34582aef1eb878cb992e9 : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/76038b192c4746f1a8160032fac9d5d9 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92d1fc016a584ff0b1fa724e171be488 : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2439d763d7954b44bc0f746cecfca5cf : SUCCESS in 7m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d42e91df3dd498dbec5cbf3e37c62fa : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b98c709310834a61b30ff3736fe93597 : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caa0daf247e44963a803b74ee2707267 : SUCCESS in 59m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84c44f88c203417483e1c5635a8439b3 : SUCCESS in 1h 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"88aa9c469752012163d75d1d9dc204aa06f25a52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 00:04:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52f2637fbe81b16fa14af1add8ae34463d712f30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 00:43:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d3d6291c8f5248e98d8dd8f0f2003568 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"09d649fc616ee7a836cae14db211137fd943bd78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:37:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37cb6d0c051459daa62bd3d1b9bfe25 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4394976c94d34f2cb05658c0d3be4c41 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b9326dcf4bb4982bbd4dae2eaf47988 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/622063ccbd2543ec9eed68f7dda8418d : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5269d552d3c4026959cbd19643e831d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d383527943f4b72996b65fb3c16c286 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ea6c113699244c0bc439ff1ab22dd14 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/abac694ee8cc457db49b2fd2ab9f3aa2 : SUCCESS in 10m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1b78518ab5a4879adfb4a88a6c7078f : SUCCESS in 12m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a618153c590404a9469bdc4fb02ee34 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9dd1820f5ac24a53b94a92dc42edef04 : SUCCESS in 7m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0f10940cb6f246abb5a206c629a3576d : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85acd82c764b4b0da98306a871a7a427 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f04e1461cbc940db8452cef137668473 : SUCCESS in 11m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2f187af836ba415dac147e96f85a1184 : SUCCESS in 15m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/65ff236ed0e34c3d97bcc06d3763b8c5 : SUCCESS in 6m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc915a657d1641f094fbe549b3c6c5c0 : SUCCESS in 41m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d29e3cf1996543879b0d7cf81e46b604 : SUCCESS in 13m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e000e0b08b8c46f9975043ef82bf1761 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/951dacfe24a24a199275a5d2b4ef1425 : SUCCESS in 20m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dc7166e3d928409f8eeb82608a2b33b5 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9693066cd2634fa9b60e4fd81cc19e80 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c6db61b915874724afd83a5dcfbfc03f : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0e0ceb985814069be67b8cecc82eac6 : SUCCESS in 55m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a19889e50a648079248ff763a6d19a8 : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"14e2512c7481aa2add64acfcea1eebc9b95037aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:37:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d992e80ac50f9a921ff778c7af941f1d2daa95d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 07:08:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca980c089e243e3be000646f117df7f : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7f96c1753f4fb287f5748dd635bf1a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1773dc8ba4a43c1b9273d43f1c374a9 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e9202b179ff4c458aaf4694db0c2ae2 : SUCCESS in 10m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d993f11820e543b7b70b9826c89e7e5b : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6fd6784020d424abe4a849ad0a3d01e : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3987bcbd22734181a063836c78d6e837 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e72bad9354f24e94a1172012141296a5 : SUCCESS in 11m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a4ee80dc0fd4279a259ccc7ff141dd3 : FAILURE in 11m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fa684e6defcb48d7a6287c39878b55b4 : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9170312d62c465f8b7f7fca3ab0b64f : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c42777f68ca4f419ad7060630d91257 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/272f4fdfc07f41a8a7c9b0ae9efdc134 : SUCCESS in 12m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bf3d1c8a43684a85bf418f71984f2eb9 : SUCCESS in 18m 35s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3e5cebe058ed4843ab3b5450bcf9fada : SUCCESS in 6m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f5ba22f328c41e99e55bf9541511ce3 : SUCCESS in 31m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8efa8fa54734331add8810304347992 : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3250ede060bc4459aa8b1e7c622b4858 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf2edf62513449e78a295c9dc7ca3e57 : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b3e43ce5496418096047dc11472481a : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd6e296959d14aabb43693e9c636b7ad : SUCCESS in 6m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/89d445f6e55b42bfbb6606b1a97f8cd3 : SUCCESS in 16m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f25b2fc3ff7141c4aa4391878392294f : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ef596a9310a4e47bc98d65a95923284 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":1},{"id":"830e0a0952f6593970c8b43d3eb4f97ec1169778","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 17:59:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"21a8894666c981db2116ae38f5309144048ea3b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 18:34:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0388d80486804c168f01c87aa1481aa8 : FAILURE in 29m 58s","accounts_in_message":[],"_revision_number":1},{"id":"aed5fad9d8765600704f303b89d0f57476f50367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 19:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7029615ae1e344a291227186876d283e : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109f54abdfa545459b2c50df403edcdb : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63894c4427f740cd8eea09dbb29674d9 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7468c11f4b14998ad9f0b3783ab95f3 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e57f5a6080fe4c1b8aeadf3ea6456a97 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20a286067a344982a3950a75f5a86d00 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/233217c48c804ec2800e46661842cf1f : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f57d3368b0f43ed8260849e155a289e : SUCCESS in 10m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8adc24038dfe41c58aa9c1afc7c83344 : SUCCESS in 10m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78a6ebcaf9d3488bad80ec558666cd02 : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e5fafa1bc254da38fb489a243bada7a : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73d000b745cb4e5a81085afd924c2a3d : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b05912d92f74c89ab6f500db6bd6986 : SUCCESS in 7m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/08faa15d9ff24b909397f9888f890d0f : SUCCESS in 10m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5e63bea4f4824f70b2d99f750643bbc5 : SUCCESS in 15m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1221833120aa4135a7ee4ea608acf2ed : SUCCESS in 4m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4118af64a920434b824f2824e6f3c923 : SUCCESS in 32m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/29c2c6e21dee476b8aab293477eeee17 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4de13078c624cb3a89336061ebe5669 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6560f442b0564ee48bc4fca6f661abd9 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/573c85891f5c4fa180ea191b1747270a : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6eb872d3a76a4c2ab3684884ceb4653f : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4236f6b3da7431b850edf7c7e8edd99 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e22703bbab384cb09fd5c1c60ce8d22b : SUCCESS in 55m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2017bc1fa96d46b187d98f858753c0e6 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9cbe8bc9b5df4f5b68e62005970cf80a23384349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 19:36:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f441a2e438c646ec9ddd4fd7a64cc22e648f6fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 21:37:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bea7db77014719a7432794d23b1d93 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2625d121e5ee45b1a7dcd5bac1e61253 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f088fb80f0004bc1b969bf708d68b231 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fedb7bc45154e7ebaf467b25f1afef9 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/886f23a7751b4d87b7dfe953b2facd8c : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd1b16f68901409b99fe63b8628c975c : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9744d6698c8488d9a4f3a6ed1c06f05 : SUCCESS in 10m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57ca4c00a1e34f67a6d235025956ef94 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13ea04cd08ff4b97b107c8a759e59632 : SUCCESS in 11m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/341545c080034b6e82589613abe1960d : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2c4d9e0a3fb4aa1842dd396ece00d4c : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49e3fac30a2d4d979423f63e9c8b72f2 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a50d8c462a074b8192803f92e17727a0 : SUCCESS in 11m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fef72245e954e878af3f635ad680a97 : SUCCESS in 13m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fb08358531484ce9b5cc39bd138ffaf8 : SUCCESS in 6m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c3e3c544e9e148e7b67017bac8908c3f : SUCCESS in 33m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d664e5ff2c034e0ca52acb7f09cf4bc6 : SUCCESS in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28a0c203767c4fc7ada9de516f11d9d0 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ec2d10f1f154a1f8c010df884d4facc : SUCCESS in 21m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9d6a2f87298246e8affed66981d2b436 : SUCCESS in 7m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55bc4c74f5e44110b6b31754540c3693 : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a35b3e8cf35e466e8143f1f188b838b5 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a160606db5b94c61a4659f97fceed198 : SUCCESS in 53m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e4761af4728447e8fe2cea3de9b0610 : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"31ac2b94c1e74888c2428cc9633a9760debe44ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 21:37:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1f4dc1b749d5f5f5df9129dfeeeaa69c3a8ab73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 21:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c3b9c10acb4c058d97a70bd30ae65b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"312e6fb40ca393e045824aa57bb5eed674b88c8d","revisions":{"312e6fb40ca393e045824aa57bb5eed674b88c8d":{"kind":"REWORK","_number":1,"created":"2021-03-12 16:53:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/780299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/780299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/780299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/780299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/780299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/780299/1"}}},"commit":{"parents":[{"commit":"45e4013e794892d234ab004be7093644fa776cfa","subject":"update constraint for openstacksdk to new release 0.55.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e4013e794892d234ab004be7093644fa776cfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 16:53:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 16:53:45.000000000","tz":0},"subject":"update constraint for python-ironic-inspector-client to new release 4.5.0","message":"update constraint for python-ironic-inspector-client to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: I6977c691931828863e19203b918231acaf64ded4\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I46eda7b2cf81049a3790786f418797a27cf53d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/312e6fb40ca393e045824aa57bb5eed674b88c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/312e6fb40ca393e045824aa57bb5eed674b88c8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
