)]}'
{"id":"openstack%2Frequirements~780301","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I0d36ded83737403fb67bb309de98a11a21eb4e82","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I0d36ded83737403fb67bb309de98a11a21eb4e82","subject":"update constraint for ironic-lib to new release 4.4.1","status":"MERGED","created":"2021-03-12 17:22:57.000000000","updated":"2021-03-13 01:26:54.000000000","submitted":"2021-03-13 01:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780301-new-release","meta_rev_id":"bdbed2b38ad02be84c12f87cb2fad4ab05ad9462","_number":780301,"virtual_id_number":780301,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 01:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-12 18:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-12 17:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-12 18:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 17:30:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-12 17:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 18:13:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-12 19:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae440d5b2bde2a78de7401d4d99f122e78af19e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 17:22:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d8edff185b9d787e906fc4f2b93b44ef7f964d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 17:30:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0bf4cb29792bec1843e989cacc41f8dfef7aea88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 17:50:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1bb32deb5e234cf98f45dc5f7996c054 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7ab24193c8c082a2f05d5da422983487f4dba2d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-12 18:13:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"752844d417fa18badea7b37b6fad308429c745eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:13:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10caf4083a04ee88be060908f7183cf : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d525551afb94d088d1f2ce0e0b83dbb : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17d7b07545cb4c31be51c7642c279b7a : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a49e5c30e404e7097b733505aa86009 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8458c85888c48748750139f32350a5f : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1de8f8565eb24bdb86e54335df113c4d : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28a5f61252644dd1b9c4ede37d69d5e2 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10377e3dabbb43c0a476126f56e31e20 : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/09cfe0923d3d4d398966f47865d75046 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b2541226f214ba9addf7c35ca854ef1 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f3c12e5b1844202b5d8c21ef4d8cf1e : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/01948ddc4e5a4e08a8d990b1ae28cea8 : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/786141f9a7a54ebf8d2b27c0b879deec : SUCCESS in 13m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a375254262b94af08b1e9dbed6e6fc73 : SUCCESS in 31m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/256cc0ae03a846008448182cc9c34320 : SUCCESS in 15m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83c3017eac8a462f977f44f5920db1a2 : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89be86778ff449cb9a51682785cdd85d : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80499d915956475e9ae74870a80a9ee4 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e47596216cd4b82a5c1862275e467fd : SUCCESS in 8m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fe9118f0b9148e59d3ab15bb7d18751 : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7ea4f1b6ba441748517db0d67bfd180 : SUCCESS in 1h 03m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fd397e453e64d75ae6d741ae22372dc : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1468dfc462fc91bfa65a99af48180f2ab07fd42e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 19:19:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"528905858d98e614dd0bf182e55b1e71dc2b0899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 21:14:47.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5977d7c4a4453adc01e9a4055dc68ebf7abff253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:26:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c06bf3096a4d178aca9c8daa477eef : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29a7fa10b7a64fef92fbde9f960e107f : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e94fbaacc6e54f0cb75ae77a20635603 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a2e58c5ff794c3fab3a09facbeb8b04 : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ec333f786cb4d07afd825ab3d8a2ea5 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/282255f3df52498bacda7e1c2c7e47e7 : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/efa823e499ac49bd9fbc4a112217b606 : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3c145c72941241a4a6cfa380f19c2b6f : SUCCESS in 10m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3a64e3cf099f4db699273e49db62d09c : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf9b741e3afc4e5db71d86ff4f1f2979 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6badefc01c1a4f1a9b18303dd1edf067 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e4f7156a5da4c0d91fcd68cbde9f6e9 : SUCCESS in 10m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e88b8c6bfc0448de8ed2e34bf1165bfa : SUCCESS in 17m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f454cf25a7114b84a35b6da9798683ee : SUCCESS in 25m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ccb369f7ed0942e19e359d6ac058a687 : SUCCESS in 14m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/802642963a3747d5814c96d3712104d2 : SUCCESS in 6m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bede0d49d074c93a1f55778002ae152 : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f78bb25a72ff40868e21ab36fffafedc : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dfa959223f94d2a8f7de467a032c6d9 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aa921ce55a80476eaa85162e045cb7bd : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1de414691ff845898a358484bc59664c : SUCCESS in 45m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ba2611e62c847c2973ebbe9be2393bb : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bdbed2b38ad02be84c12f87cb2fad4ab05ad9462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:26:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38","revisions":{"0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38":{"kind":"REWORK","_number":1,"created":"2021-03-12 17:22:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/780301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/780301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/780301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/780301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/780301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/780301/1"}}},"commit":{"parents":[{"commit":"6b212a450e52797f124defb9442bf359e8f19573","subject":"Merge \"update constraint for os-brick to new release 4.0.2\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b212a450e52797f124defb9442bf359e8f19573"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 17:22:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 17:22:55.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.4.1","message":"update constraint for ironic-lib to new release 4.4.1\n\nmeta: version: 4.4.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Arne Wiebalck \u003cArne.Wiebalck@cern.ch\u003e\nmeta: release:Commit: Arne Wiebalck \u003cArne.Wiebalck@cern.ch\u003e\nmeta: release:Change-Id: I02a1da01258e2b7cec03b4131ef1c1025b748d82\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0d36ded83737403fb67bb309de98a11a21eb4e82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
