)]}'
{"id":"openstack%2Frequirements~780397","triplet_id":"openstack%2Frequirements~master~Idfb40ba12893d75731542044d79532ee27c2a499","project":"openstack/requirements","branch":"master","topic":"anaconda-deploy-driver","hashtags":[],"change_id":"Idfb40ba12893d75731542044d79532ee27c2a499","subject":"Add pycdlib to global-requirements.txt","status":"MERGED","created":"2021-03-13 02:03:38.000000000","updated":"2021-04-07 05:31:36.000000000","submitted":"2021-04-07 05:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780397-anaconda-deploy-driver","meta_rev_id":"db711e9d9b477631d13cfab7e90b44347b7d19f9","_number":780397,"virtual_id_number":780397,"owner":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 05:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-05 14:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-04-04 18:08:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-07 04:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-13 02:07:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-13 02:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-13 04:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-04 18:08:28.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"cacebc14482166fc9cd4085883d69780223d000a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-13 02:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"801efca88e91271d41973d74f646b9905023a589","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 02:07:59.000000000","message":"Patch Set 1: Code-Review-2\n\nDoes this need an FFE (also, some of the upper constraints stuff will need to be fixed, pkg-resources removed, readd setuptools, revert the tenacity change, revert other openstack package updates).  Is this needed for the wallaby release?","accounts_in_message":[],"_revision_number":1},{"id":"9b1a387e3dea3637bd9c98b839aa7f6870d7689e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 02:34:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/205588d8511444e780708fdffa856f4e : FAILURE in 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b18d9cf9a4eee90f9f5db673219ee94d21f52f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 04:12:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed60debea26b4a88b60ae727eadadf54 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b0d3222da945ec9933c4b286971a00 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/168d6a46bd5646baa465ac152f51a4df : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ada915f298cd4a00a2066f467df9441f : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c9fb0b835d542c5b28e6d88b9e86fbb : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d030d077038b438fb6632d4da6fe26bc : FAILURE in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c3504f681f9649f7b039cf8848c2c5f9 : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da14b40377ba45a98f907e2d99c4f6b6 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81b32ef496dc42a5a36065994a80964e : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/910697086098458b98a8d941defbc512 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/963df521727c46c39e1585bfa1c8734a : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d50554585ae74559ade85b8e482bce9c : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa818e9d43504f16abc5933a2a459803 : SUCCESS in 6m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b2407d0b1fb47879f188086251be4b4 : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0a569006bbcd4170b556b95722aabc3f : SUCCESS in 14m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/35275bdd285d42788dc0b82e803df46d : SUCCESS in 5m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d696d9b9b9547a6b683ac319efef07f : SUCCESS in 28m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/295fd76431ff4deb8972487b24fa0064 : SUCCESS in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0daef6db3904af7bdee230090c088af : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6669ef5231fc41d0b76ef0e6464aa829 : SUCCESS in 17m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca621fc2d63742b285e78f49edc686bc : FAILURE in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c546e9e0ca694415aa1a6b3af92fa2af : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4962a4bfdef046e4a922e48d977f62b6 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9da44693c6874435b963deb3aaff2ec9 : FAILURE in 1h 01m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a33a51b793d7401d87f6b96d656f09ae : TIMED_OUT in 2h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"19ae120dba034f7c409a08236c848ee31dab6eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-13 07:05:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab97c3d099cbc13fac3a58421b4810f6402c3b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-13 07:08:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a0bb4b86b67264a4ab7cdfb33b401d8012fb5787","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-13 07:09:37.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 1: Code-Review-2\n\u003e \n\u003e Does this need an FFE (also, some of the upper constraints stuff will need to be fixed, pkg-resources removed, readd setuptools, revert the tenacity change, revert other openstack package updates).  Is this needed for the wallaby release?\n\nThe feature in question is one of thw wallaby priorities for Ironic project. I think it will need a  FFE. I have reverted the removals and version bumps.","accounts_in_message":[],"_revision_number":3},{"id":"844834bc6e613ef45a13a91bf218ac2dc5d602de","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-13 07:10:57.000000000","message":"Topic set to anaconda-deploy-driver","accounts_in_message":[],"_revision_number":3},{"id":"cee5219989c92df650f2854ee1842db4972a2350","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 07:44:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fce269124cc04f37ac0b163c6f1c5b43 : FAILURE in 31m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f1fea25f500d37b32ad367d486f053749ada055c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 08:18:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f7e50080584541a240bc11f9d60d70 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be418c27995c4500b74fc70994a6bebe : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7526fdcd86e4bf4b6cafb51f74acd8f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d7ae66bf2b4c06923deff7c4f5bdb6 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3833108322a4e0eadddb435a0f95d89 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11beced6599b4f34b5da54b8d4ef06e4 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ecaba8a6a884a60a8dfdfdffab40cd3 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d8e375f9ab743b5acf55fdaa069b8fb : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/291ee181b96f4296bbf14bbef3b2a1d6 : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98b090eb7c3a466f88012f8729c9fe5c : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff7011df0f1342388aa6847204388491 : SUCCESS in 7m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1ca9eb3e01c04171a023db5dbb7104b0 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2652facb2cd4b27840c0d15bb978b83 : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73bb1350f2fb4f81ac3eabdf2ee85acf : SUCCESS in 11m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d013209df9d4477e8977661b0dfdda5a : SUCCESS in 14m 29s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e01c3fea1c2c48aaab5be89ce504703f : SUCCESS in 6m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fa128a4d604421aaa144ffdf4fb812b : SUCCESS in 33m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b01111784a144ec39b56c76030793a76 : SUCCESS in 13m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/245a9fa9822b4e1ab01900d154d98b27 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fa3a5c1e90f465b8cdb35b0dffd9f73 : SUCCESS in 17m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0346392dd076492da32c44eea86dd09b : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f667c9034034dfca46308b4c6f4b82a : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f6ad8ca8d9840f0bf9abc6851fa76c9 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d13a328a69d14624a1ac28b791b60223 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83f9556f38f44a2ebaa692e6c635a1ab : FAILURE in 1h 08m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7ea4719ad468910e15a540eb8dd94cf0a9996070","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 18:00:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e60a2a93e9abe43fd0956e51dca8ee86351fad1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 18:34:48.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7cf5180175f4ebd8b851bd2891275dc : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":3},{"id":"5248476738b9309a48e456131ee5a7288ff07105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 19:34:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42692bb51c5f49f7b490b1cf342285d9 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2af27074764c9ea6ed8a677260fb47 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a8adbb105743949017e041a90ec06f : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad664f6443c4604a4f06817d302ac88 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84c06787b3734d728fed8c201d80142b : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/affb0d13d6244eccaa355205903abd59 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e9d59ccb6ea4d57afcf724f028326ea : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77f0c2cc11c54bc781eee53d74236f7f : SUCCESS in 15m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50a736124135401f983a1125cd91a1de : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c22b039f5994f378294257a741a48d8 : SUCCESS in 12m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/82baa19c380b4df8bc340ca838a9861c : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/56cd1a850773454b958dc51768ec4e96 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86a5ec112452480aa0a6e4f164d4b7cf : SUCCESS in 10m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/afef7b922f7744a8820e22781c3131bb : SUCCESS in 10m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1733cfab38649e29314688fdf2d1073 : SUCCESS in 15m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/94875e814c6341d9930417e05f103d12 : SUCCESS in 5m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5ccb1f138944c349b5561227205db17 : SUCCESS in 32m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/025dfda94ea8401d9813d6713175fcb7 : SUCCESS in 15m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/441b8beb5859417ea3a1cc784074ba25 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0d4817f3ebd49e5966540da830aa40c : SUCCESS in 18m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8a7e0cb86391488e9bce4155e8f4b902 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05bfb237a77c42aab574e0df3b988845 : SUCCESS in 6m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b8423f47a13e42628faef857756d35f6 : SUCCESS in 15m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a49c2ef80bf41fc9c4ae191c186dd37 : SUCCESS in 56m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d40bdbb246c2400793419d4bcd55963e : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":3},{"id":"26e02f4a41fb9b9b62be641ee6b3c8c50af7f0a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 20:54:21.000000000","message":"Patch Set 3:\n\nplease let me know when you send the FFE request to the mailing list.  I don\u0027t expect any issues, but need to follow the rules.","accounts_in_message":[],"_revision_number":3},{"id":"96156d248ff89de57d147516b3cb4ca3645c1d2c","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2021-03-18 17:02:19.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e please let me know when you send the FFE request to the mailing list.  I don\u0027t expect any issues, but need to follow the rules.\n\nI had a chat with ironic PTL. We probably won\u0027t be needing pycdlib for wallaby. So this can wait until next release. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"90559d10642b3e16a1c61e87f62c1b5c6943f214","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-04-04 18:08:28.000000000","message":"Patch Set 3: Code-Review+1\n\nCan this be re-evaluated now that wallaby has been branched? Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"e6d80e2e0914c55b9ec326dead3761b823c09f73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 14:45:13.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"b8ee8e4461178829d69c4580dc8f82673ac44ae5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 14:46:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"02c99481671c0ce33e0b3746cb53875b7914394e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-07 04:09:54.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ab7374d172ee94bdde6948b06b92b19c3a33c55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 04:10:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e400005826575acd56f5ec5b0b944c31bb31b35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:30:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c843691142450697c09aa264683122 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9911a317f44277bb0448c2423056e7 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b5d3380bf041e0a1737112512b658f : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be15f5c7334d4137a52ef94bde2b0bb6 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/830e65672bf74c00b6709e1b030035cf : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38fdf604d9a04fd39457fe4111f2ac0e : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b363a5976114c08bda4f733709875b9 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f2dd5892fd7a40cbb36f065fddc6a907 : SUCCESS in 8m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d4142da090647a592aa0192664a0197 : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3ba946c3030440ba8e2c696dcd3abcf3 : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6712633d68ed4721b1ce8245b8be8dcf : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb548e6b8ff5463ba3b93d10d90213e0 : SUCCESS in 6m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb2580510cff425c82212a98e55482a0 : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/14a2ab43e71e49629e255da476bb22ac : SUCCESS in 16m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/afc228fde7e0449ab9e55e7b28240ea1 : SUCCESS in 5m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6afc6bacf924e5c8916e7c2b66b8a89 : SUCCESS in 26m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e8676d4e43c54b5ea311ed05330ef550 : SUCCESS in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/268877f1ab4044008c1fc9e8b6f97328 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb1992e1d68e405291dc96168fe78b55 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa09fec6f26d4719a6a8fb54513b6d0e : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26d6dcf024c64a7b96dbe940d3f58f23 : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73aabc5210734e5f81b1efe8c26a52f0 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15965afbea0b4c8094fa281af0faad3b : SUCCESS in 58m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e5fd5d4503f4045be8f213362daf47f : SUCCESS in 1h 19m 03s","accounts_in_message":[],"_revision_number":3},{"id":"03cd23d2ef0114d4a8034115113ea4dbe9e923b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"db711e9d9b477631d13cfab7e90b44347b7d19f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:31:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33aaff8f5a494a9dafc574bdeb558d1f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"607b5a5170f0534811b1526e35ca9fdba6f92926","revisions":{"7406aa45b4bf09874be319077eb6c19a91ab8ff3":{"kind":"REWORK","_number":1,"created":"2021-03-13 02:03:38.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/97/780397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/780397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/780397/1"}}},"commit":{"parents":[{"commit":"810aa65b8d47d876af227295a433a39d736ed90f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810aa65b8d47d876af227295a433a39d736ed90f"}]}],"author":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 00:29:08.000000000","tz":0},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 02:03:27.000000000","tz":-480},"subject":"Add pycdlib to global-requirements.txt","message":"Add pycdlib to global-requirements.txt\n\npycdlib is a pure python library used to parse,\nread and write ISO9660 files. With LGPLv2 as the\nlicense it confirms to openstack licensing guidelines.\n\nThis is an actively maintained library. As of this\nwriting the last change to this library was 8 days\nago and the first commit was over 6 years ago.\n\nThis library is compatible with python3\nhttps://clalancette.github.io/pycdlib/python-compatibility.html\nIt is a well documented library and the documentation\nis available at https://clalancette.github.io/pycdlib\n\nThis library will be used by ironic to read config\ndrive and write them to disk on the filesystem.\n\nChange-Id: Idfb40ba12893d75731542044d79532ee27c2a499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7406aa45b4bf09874be319077eb6c19a91ab8ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7406aa45b4bf09874be319077eb6c19a91ab8ff3"}]},"branch":"refs/heads/master"},"14240711dda63d916adcee629e053a88b1cab386":{"kind":"REWORK","_number":2,"created":"2021-03-13 07:05:23.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/97/780397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/780397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/780397/2"}}},"commit":{"parents":[{"commit":"810aa65b8d47d876af227295a433a39d736ed90f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810aa65b8d47d876af227295a433a39d736ed90f"}]}],"author":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 00:29:08.000000000","tz":0},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 07:05:16.000000000","tz":-480},"subject":"Add pycdlib to global-requirements.txt","message":"Add pycdlib to global-requirements.txt\n\npycdlib is a pure python library used to parse,\nread and write ISO9660 files. With LGPLv2 as the\nlicense it confirms to openstack licensing guidelines.\n\nThis is an actively maintained library. As of this\nwriting the last change to this library was 8 days\nago and the first commit was over 6 years ago.\n\nThis library is compatible with python3\nhttps://clalancette.github.io/pycdlib/python-compatibility.html\nIt is a well documented library and the documentation\nis available at https://clalancette.github.io/pycdlib\n\nThis library will be used by ironic to read config\ndrive and write them to disk on the filesystem.\n\nChange-Id: Idfb40ba12893d75731542044d79532ee27c2a499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14240711dda63d916adcee629e053a88b1cab386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14240711dda63d916adcee629e053a88b1cab386"}]},"branch":"refs/heads/master"},"607b5a5170f0534811b1526e35ca9fdba6f92926":{"kind":"REWORK","_number":3,"created":"2021-03-13 07:08:13.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/97/780397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/780397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/780397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/780397/3"}}},"commit":{"parents":[{"commit":"810aa65b8d47d876af227295a433a39d736ed90f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810aa65b8d47d876af227295a433a39d736ed90f"}]}],"author":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 00:29:08.000000000","tz":0},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2021-03-13 07:08:06.000000000","tz":-480},"subject":"Add pycdlib to global-requirements.txt","message":"Add pycdlib to global-requirements.txt\n\npycdlib is a pure python library used to parse,\nread and write ISO9660 files. With LGPLv2 as the\nlicense it confirms to openstack licensing guidelines.\n\nThis is an actively maintained library. As of this\nwriting the last change to this library was 8 days\nago and the first commit was over 6 years ago.\n\nThis library is compatible with python3\nhttps://clalancette.github.io/pycdlib/python-compatibility.html\nIt is a well documented library and the documentation\nis available at https://clalancette.github.io/pycdlib\n\nThis library will be used by ironic to read config\ndrive and write them to disk on the filesystem.\n\nChange-Id: Idfb40ba12893d75731542044d79532ee27c2a499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/607b5a5170f0534811b1526e35ca9fdba6f92926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/607b5a5170f0534811b1526e35ca9fdba6f92926"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
