)]}'
{"id":"openstack%2Frequirements~780406","triplet_id":"openstack%2Frequirements~master~Iaf94677e34d70da82dde94fe831b3a07331ad87b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Iaf94677e34d70da82dde94fe831b3a07331ad87b","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-03-13 06:25:03.000000000","updated":"2021-03-18 03:52:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3562f8a8449a9fd4f4d0ab1a72d5c050953f3b6f","_number":780406,"virtual_id_number":780406,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-17 07:24:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-2,"date":"2021-03-18 03:52:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-13 06:46:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-13 08:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 08:54:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"475bf6c1764ecf3e88ab3def40c8ba33ee299c64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-13 06:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ce49c3bea3551382cc76b77a08f3f07d81dcc0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 06:46:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c9059e9cde864e8895e9c1e4ae8fe7c0 : FAILURE in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0d25ab01bd636c516c35b4de22885c52ff6daf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 08:36:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad6691da6734cc0bf7f2c30fad90102 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b150a902ec84a76b2cc16dc1a1cf319 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bff376e1b1044e988bbc21220a84768b : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f881c5ed44964f619d7afbb5dd65c715 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c5fb545fbe242989f07aa59ddb6a14a : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b8d47ed7acb47c8a703c67801c41312 : SUCCESS in 7m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98215ec2f8f94195a27af5ed65b0247f : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f0c3d04861a4090819a59fc2508582a : FAILURE in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bb14e6482014dd89cb67cc0d319777b : FAILURE in 6m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3dee417bbe1148128caabec1dd5ab828 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0763a9088f4041a98236a8355a70ff4a : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/756d2ca10a64435a90a7c8597022d81b : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef4df3fcf79441e2ad406ddb4aac5fc7 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6b82ad25953c4bad8a4a37b9a89dbbfb : SUCCESS in 11m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c566317bf68c48cca0eb9c9b7999026f : SUCCESS in 43m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/405e567902d74001a440163a621c15e9 : SUCCESS in 5m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a4deb74e2edb475ebaea8ebccc107c5b : SUCCESS in 25m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd4fe9aeff3a4ea8b5125d3b1858b370 : SUCCESS in 10m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/596b6978070f4e799564cbbe2b432c33 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7035145799b94ce6a6c5fcf83d525545 : SUCCESS in 14m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a8024c10bcac4ce990e2135d074d8a91 : FAILURE in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae3f211115fe489683d352c0db8f142a : SUCCESS in 7m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1d3719115bc340388c8b4f2f1f7914ff : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6b2e6ac4a774f66935e517f0a921743 : FAILURE in 1h 04m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/716ac0382b764e8db1267d3a5304d4c4 : TIMED_OUT in 2h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"734a986d2b97949b9f39c4e7d0c0bb2b3ed3150e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-14 06:33:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7d178c1719876125e6e2ec647a189d57875d619","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-14 06:46:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/492b7d5b8fbc435b9e20e9451d89f411 : FAILURE in 8m 09s","accounts_in_message":[],"_revision_number":2},{"id":"cc8145948143eae0889a1ba0d24dfeb5c812acd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-14 08:40:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d12f73dcb44093a7d9434008ab1925 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25be8588889416aa2343f338a75d181 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b1645b940d64623a06d139199397dbc : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a37bd39b88c4965ae28112f1d1dbff7 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b92cfbb7bf04fc9a0a0efce484ec425 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14c6f4bc1c06408a9890fef5bf4a8603 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77b9d761b6524e8e9e8c0875559fd481 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a99b147ee4e141e5abc249d820c32c18 : FAILURE in 8m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a7062d4bcc94f00a2dca37568a0ab32 : FAILURE in 5m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89a8ab3276a6462a953e37df679a5622 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d0065f63536b475ebe1e12283fb8fc34 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7be2496802fb457ab9b373cccb17f8e2 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d595eb1462d484f9fea658c47e3cb77 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e5725b84149c49cfa61f04166d597412 : SUCCESS in 9m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ff15ebe15cd4ea7bf636062bda51919 : SUCCESS in 14m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2decbfce1d5542c3b35aa459ff4f394d : SUCCESS in 4m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f2806e67abdf4593ba4ce66762f462d1 : SUCCESS in 34m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/092525de56a34176948891aa68bafdc7 : SUCCESS in 12m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2eb778fbde07411ca4d59e9e63f79694 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd9ebd82f5bb42e687699ae52b7acf2c : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/72e312ba23224658aaade694b4403b3e : FAILURE in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de813e6e2331401fb1053b83f424a91e : SUCCESS in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3de35bea7c6340cdac506cb3cbb0fd95 : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42095148586b48289f5e816eb8ce9c9b : FAILURE in 57m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c5ab0c286c44bf7b4824f892c8fa44d : TIMED_OUT in 2h 04m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8750c6c6609ecf704a6b5993d23a68737bf41147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 06:20:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a1dd77b4e50e3bcb440709eaaec528a92cb8d8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:36:55.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6cca3baca5a948569e62a241cf149d57 : FAILURE in 9m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e7280ff60dd73b1943cb1ebf810d41e96f2d6a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 08:45:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6877295dcf488190eb90cd1089c28b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415d2231d9e5405f9be486162fea1a63 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f34c539a4e400d8c941263165f2ac2 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1b3d969efaf4a9a9a697ae3241e22b1 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ad597e079744223bf4ac469bcdd3196 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f339004e814c4555886574764049e36c : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45779e5c816a4a4c8e4b95a219de6ff0 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8980797ce1e34dec918974eb1db55717 : FAILURE in 6m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0a2717e41ee94c8a8433a578252ee8ba : FAILURE in 6m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70b1342b02ba45b89019dd0ffe92fb2a : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fa8fc65b199e469ca9e3342405a39f3e : SUCCESS in 7m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/769ffaa9efa2449ca18265ab08f35c8e : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87c6b818e5bb48219bf5d698cbd9b60f : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a550c7948842485dbc7f38b0472637c6 : SUCCESS in 11m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4718fc845ca74ab39fd390301ea2cf0a : SUCCESS in 15m 43s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1b2b980d01554a858ab3330fd7a5c9ab : SUCCESS in 7m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/16ab049f4cd14845a0e1f116a9c596a6 : SUCCESS in 30m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/50d99aff46354ae3995c63aadf7e8039 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1aa3acd30af64761aac20a4527b0d016 : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98a3b0d3d06347cebe01bf3231d1c6c5 : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/551a3ea6bcfc4123be78cea13f824771 : FAILURE in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c07ae2a5862745d482d53c56f851684d : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14ae0481eed3430b8cdd5b2816e76d9c : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/852a1eaa41b54f2485329d4f18293ac7 : FAILURE in 48m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6134a475a424d9d9b440a71684224d3 : TIMED_OUT in 2h 04m 31s","accounts_in_message":[],"_revision_number":3},{"id":"40d917891e3d2c88239b7840c1fec4b13ccd3d8a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 08:54:00.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"a2124323b0ba9f2f5d74f11cd31a930f422798eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-16 06:30:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba213042e76bb7d3337f0ad62c3d2bb2fcfea132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 06:45:49.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b18b391311d64980a697ba0bb51abfc2 : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0e8f2c6e054b840f19503c3cf8d25820b909edef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 07:10:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572301f8aba14efb990fe0f634e88e17 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bffb3276ef6548b7aa484d597c1e11be : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f58c4bb3ad4f7799ebad3c77ead7e1 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d96f37d1484919970bd09535794bb8 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58c4ce91f7d74d98bf654c6d8fb2350e : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6179b7c6662743cead416dae2414dea9 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bd297faa75b4334be352e70b0f19c3a : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5f9743fba5a4519a9b4c325e9ce82c3 : FAILURE in 5m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b82946368da34927b5c5834307583d6a : FAILURE in 7m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/392460d98a544998a7b5a6c5dbe1f785 : FAILURE in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f7e889e8270457baf7388f9a9154cfb : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/346bb25ce69a41e9887e615a34fd333f : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecf2a4620fd94eda8051bdb1f79d81d8 : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb054531071c42b6b60ca3be028726e9 : FAILURE in 10m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/489459084f80412780607b7cfc6c6b5d : FAILURE in 12m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5ae860cbd9534a40aaa20b55e5243d18 : FAILURE in 6m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d3f8a03fe8d4f38b3074ff55d33d01a : FAILURE in 31m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/929b015d6b2c45c8a497ad55acd6c17e : FAILURE in 13m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30495f835bc24797ba1ca1fa8e0af46e : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da7cecb45ed74dfab1ec5270c16fc5f1 : FAILURE in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e0ef81b986f244bda0fc4df850233c5f : FAILURE in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e559509f51d24c489b5463244a590d1f : SUCCESS in 8m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8d98cc6cc130498ca9b437e92447d695 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/797a60fd8c1a47689bf99353752b65e3 : POST_FAILURE in 28m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02164b9a29e84eacb42b2967c3e96458 : FAILURE in 26m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d01f702738f57c95bf062713552247cdb23a4913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-17 06:21:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a886b963f589202a6d925671a434c943e8b1ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 06:40:40.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0ca52a39e3641d1ae11d9ff597365ec : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":5},{"id":"28d9e781cd420fc67bd918ca1a3c9d514f1466f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 07:24:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9946bd553c4a4897c980f3a7cc56c6 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70533b2ee1eb49c4b2a34b3203a56968 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58e883da6a8d4e858af01691a686e6e4 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2c45100f061410fb12b3ffcb5d4a8f3 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0f36dab1d640e297db278b84b96b84 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2691302d027f423aa38a41c02082456d : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5dec48e48b8345b8b335b4c0b751d045 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d56e3fe027f649c89c8595875adf2774 : FAILURE in 5m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e1ccf009578410cbc2ec40d81fe9ac9 : FAILURE in 5m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d74c3e8954148f88439e394631b7dc8 : FAILURE in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef277b9f214d447ba1e8da1f2c3cfd6f : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/242f03cae7a1461f93d972a3c1fb6140 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92344fd0541e45e38f4b8547b0e81a5a : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0c532e78da24587885b869548c27f04 : FAILURE in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/636fbd9971084424871f49b86f56fe4b : FAILURE in 11m 21s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/efd9b8136bfa4c39ba85030b0d56f439 : FAILURE in 6m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d761e69f0c2342539a4fbd840a0eca9e : FAILURE in 11m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08cbc423b1594a4db5dc993fce5b5d72 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d93971bdad84a828296394e8f186fc4 : FAILURE in 21m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9c14633afe3544099e053511b3f0054c : FAILURE in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a710691698047a28873473ad0868b93 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31554dd9a5864aa3bbc8de61d03f5b44 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87cc015bf77b43a5bdf3f853630ea508 : POST_FAILURE in 30m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ecebfe687399488183bd4f25d3eafb60 : FAILURE in 20m 32s","accounts_in_message":[],"_revision_number":5},{"id":"1a9b0aabe2c4414ed1c0f5e1d77c3c4bff0f9b1f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-18 03:52:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8ebb9502dcb59a5ac964ebb502062628c85f61b7","revisions":{"475ce38a3d92ca73166ce9ea438fb55696077fa9":{"kind":"REWORK","_number":1,"created":"2021-03-13 06:25:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780406/1"}}},"commit":{"parents":[{"commit":"cddaac1bdddef7e8510a5139e0367bbe27915dc6","subject":"Merge \"update constraint for aodhclient to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cddaac1bdddef7e8510a5139e0367bbe27915dc6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-13 06:24:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-13 06:24:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf94677e34d70da82dde94fe831b3a07331ad87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/475ce38a3d92ca73166ce9ea438fb55696077fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/475ce38a3d92ca73166ce9ea438fb55696077fa9"}]},"branch":"refs/heads/master"},"74139372f4face7586e958f2583e324e1dc89ef2":{"kind":"REWORK","_number":2,"created":"2021-03-14 06:33:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780406/2"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-14 06:33:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-14 06:33:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf94677e34d70da82dde94fe831b3a07331ad87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74139372f4face7586e958f2583e324e1dc89ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74139372f4face7586e958f2583e324e1dc89ef2"}]},"branch":"refs/heads/master"},"3756d3bd7a8d55bbc688ac0721c63523d5c79499":{"kind":"REWORK","_number":3,"created":"2021-03-15 06:20:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780406/3"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 06:20:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 06:20:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf94677e34d70da82dde94fe831b3a07331ad87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3756d3bd7a8d55bbc688ac0721c63523d5c79499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3756d3bd7a8d55bbc688ac0721c63523d5c79499"}]},"branch":"refs/heads/master"},"1977496c20b1f0bca607acdc4ab1517a34b52ee5":{"kind":"REWORK","_number":4,"created":"2021-03-16 06:30:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780406/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780406/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780406/4"}}},"commit":{"parents":[{"commit":"c4279a3522ef251b0426dd0a0a5dbc70ec009081","subject":"Merge \"update constraint for python-barbicanclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4279a3522ef251b0426dd0a0a5dbc70ec009081"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-16 06:30:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-16 06:30:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf94677e34d70da82dde94fe831b3a07331ad87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1977496c20b1f0bca607acdc4ab1517a34b52ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1977496c20b1f0bca607acdc4ab1517a34b52ee5"}]},"branch":"refs/heads/master"},"8ebb9502dcb59a5ac964ebb502062628c85f61b7":{"kind":"REWORK","_number":5,"created":"2021-03-17 06:21:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/780406/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/780406/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/780406/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/780406/5"}}},"commit":{"parents":[{"commit":"29eac2463669cd7ea48c640715f7c518991d83aa","subject":"update constraint for neutron-lib to new release 2.10.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29eac2463669cd7ea48c640715f7c518991d83aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-17 06:21:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-17 06:21:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaf94677e34d70da82dde94fe831b3a07331ad87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebb9502dcb59a5ac964ebb502062628c85f61b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebb9502dcb59a5ac964ebb502062628c85f61b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
