)]}'
{"id":"openstack%2Frequirements~780504","triplet_id":"openstack%2Frequirements~master~I08740bafab73c7d5cd9a0f621cf8b25cd61bbe9e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I08740bafab73c7d5cd9a0f621cf8b25cd61bbe9e","subject":"update constraint for python-cinderclient to new release 7.4.0","status":"MERGED","created":"2021-03-15 08:57:35.000000000","updated":"2021-03-15 22:38:16.000000000","submitted":"2021-03-15 22:37:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780504-new-release","meta_rev_id":"2a20be5e39d9b068e61a358b5b81c1932d0ab9c1","_number":780504,"virtual_id_number":780504,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 22:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 09:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:15:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 09:03:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 09:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 11:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:15:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fee67d2652fbe68974b96b89e7a50a4581a864df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 08:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6737502b40d61c710b889b5c76ee15374e5b7242","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 09:03:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0151702a87c18037a005e26552138a23da376a8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:37:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8fe9ac9e80324f32a1fbdf9be3b17033 : FAILURE in 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b2db19e4ee1c99e0831d8153a91f38d507a726ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:32:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf60fc99a6f428facf55458e302cda4 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44465a76c90f468e87bc4d803568d32e : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c66410ee4da7417d9566191063b5c4b9 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/617e1a74b9c24d0fa577110e755764cc : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/580e55772a5c4c8b8ac8ad03b83858ea : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebc07b681da0441a98658e08986722ec : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a717747fd09a465da8b71c879e573669 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75a3a9144cdb453c9eb3d29d40e7ee28 : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4556468518bc441990c09a514ece462f : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bad49a92d0274d83865bd969a2ab556c : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ccadaaa5d9ea49e285e4dc2c7a8123de : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/819ca2f36b43468fa5e1f06d3f159c62 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf11402a126a4309a9e99281f2e17428 : SUCCESS in 11m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4036b9177df341a9be296d09bed796ff : SUCCESS in 11m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b08198d28888486885d8955aeb9b2bff : SUCCESS in 16m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3731b27eb9af47699dedf2ed8803273f : SUCCESS in 7m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3aa2c85c2fc14c488fe562bb5ca3cab3 : SUCCESS in 34m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2b2aba67b66048f68dee518f34490d71 : SUCCESS in 14m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0bf509c920074a5883f2f89062fc1662 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c8dbabec04c49119a134b4179099604 : SUCCESS in 23m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d5fd72291b45422b91bab51b1f1560f1 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94a9ff0e33534934824a77350418838b : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8bd46a22cbae42bcb53db2812cc09799 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fabac5b5ed904b4db34afb8f25df9560 : SUCCESS in 1h 23m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fb2e86f39584b39991c91acf8fd5536 : SUCCESS in 1h 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1f9b650d4e2571e001de5c75cf98a0da3f530631","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:15:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7abb272455f38af68141603c47f6f6838fd8d3d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:16:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b6e96ccc41caeffe3f189ee0886d62acb50b91e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:10:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d0f3c8a3bb434c95b6ee470d35a9df : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978927d01d0b406fb5df7dd5e618ee0f : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ced764b7bcc4abd8d945b5565ce5a62 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71447cfe13c44b4aa49ed7b53a7e513d : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c18fa84cc8c94a08aba63120401bf6c1 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0329a7632f7c4e9b88d8c2cf49ecd236 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/605385035f5a46d08777c71d445f92dc : SUCCESS in 12m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/466f097bc9124e82aedb28b03856e474 : SUCCESS in 13m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5afab284f38d4ca2be2f588f98fa14ec : SUCCESS in 11m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db08750f5a854d38b68fb45d0ee950c8 : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92d7d22b292a4308beb0221f2d312b3b : SUCCESS in 13m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6b030d0a93b48ef9b015879c50e6904 : SUCCESS in 10m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e15b2488c42446db9e7efb8ceb055fd0 : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f15cf3722ab74f7ba52aea1e470818b2 : SUCCESS in 15m 24s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c565928b6f644be19c815e531e2d1a61 : SUCCESS in 5m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f1e180fb3f2a4329a9b1edea5acac418 : SUCCESS in 37m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9af75509255b4ecdb8bbe97a133142fe : SUCCESS in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/409592526399430594e110a0111dd4de : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8f994f6749f44adb19eaf17fa7acf38 : SUCCESS in 21m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e532af74b80a4f408a8dbb1593b2c65d : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98041b91227a46e5a7dff5fe9d76b75a : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/57e10f7e67da4f77bb7fa069c722a389 : SUCCESS in 15m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f6e5369ae6f42f08cbdcb7e3d846fac : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/198ee85a022243b3983e62e3bbee17c3 : FAILURE in 1h 29m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6b98cfd18ff6abf9fe9df0ea0281431c879c5df4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 17:40:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5da4acc4b41bf7bb1ecf8ad83a7372095d1482f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:31:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/207b94f08b494702a581efacbe9f91d5 : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1885042d48fbcfd111f8148402400c9ed86dea98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:53:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d544abe90843442f9c79d944a1a514bc : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e820157586d84882b57a87a2c15a683a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/928e0ccf4e554d9ab99308888cfef602 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5913d00037a42bfb7612ddb966efb5e : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eddac97ac084f5087417325f0f06c8f : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91c8dc53a6d64b81a36755f2a8e43c39 : SUCCESS in 7m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db81172cdd7a456ca2c62c9d3e020d04 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ce582ebeb61412f8275285f522bae4e : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d09a26848324d27ab435c2a0acf0619 : SUCCESS in 12m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/841ea20a07514f8093e8825c5635f78c : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/879c48d897504a7e9cf4bbb41a73d9bf : SUCCESS in 8m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0a96e3dfe8bf4641adfb3009f4cddd43 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fe9b184705841398896d7c6bd1750a0 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0e19c4138fa446bb21dbd83cd769def : SUCCESS in 12m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95c2def3902040c49a72075295c4eb08 : SUCCESS in 20m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/281e13d18aac4293b95b1418c25c0e07 : SUCCESS in 6m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8133d3c4fa5419cb61426ec582d6695 : SUCCESS in 31m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/639c1c69de214e7891374f22af6c7f9c : SUCCESS in 13m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7d7e37dec1b4903b847daf5133173a1 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/506f746b3d664b40a308f95e60375d89 : SUCCESS in 25m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cec334f3bddf4ad38dd33d40098e4d49 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4254a8c3e1e41f7aae044116c55ad6c : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6963b48b299a46a087ba6765ad03d54a : SUCCESS in 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/625c48cf8aa14e1c832921cfb6abf5be : SUCCESS in 1h 04m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13eb9ee82e4346788f4dcfd86403fa20 : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1c12ce369d184ec8702a449e592bb029249d12a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:54:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b48414f12459335cd167eef3b524e6551b599c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da049af7ef314cd4b2aba1d0c5769f34 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348abbade3394133b01d59583047ab30 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8ccf4af3dd448fa15d0a0f8428bcaf : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/088de022669c4a47b492b2bda8e491a8 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f7f1d4c96774ebb9bf7386c8e87c413 : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2f16dc5afbc4af2b3d85566ce619a46 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83a565d045b14ad1b2d3a252f5fef23b : SUCCESS in 11m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e18451aa44f0470692b80a6633debe17 : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5eeb2b0f4b3e4300b3868eeecf567ee5 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2dde08b8e1ad431a9d9a25266221ab72 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ef6c0d2fd274f9abd99304bceb50303 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c46a00aea62499484e7e23e521f93d0 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/793aad7ed7cc45fbbf71d8f6d1c12c14 : SUCCESS in 10m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b2100d3d95c94395a7477c9e60ed20ea : SUCCESS in 14m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/cbb85e59dc0e41c7ac1d05d31a75c58b : SUCCESS in 4m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5461814f081c459580068441fb11deb2 : SUCCESS in 34m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a47915e88b3f4439ac575ba3983da44d : SUCCESS in 14m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f01e52bca1de4ba4989d22298ef42e31 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8e48d4d2cb5452d8d695ac5b80df0ca : SUCCESS in 21m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a35ce923d75a416b914063efd612f03c : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4296417c73e441ba0a6cb3cdd9c3128 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb62919b5f254ef8b3ba2494ef12cfac : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a92d1a2756274d15864dad222024e8e5 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6163d0825ac140759954cdb98db71788 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":1},{"id":"75b70b5031212f5d54c4f37b3a9ac1689ec08ef2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:37:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a20be5e39d9b068e61a358b5b81c1932d0ab9c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af011034627e44b7907d29d6b3a4c6a8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dee4a1e255bae09a52bf11f6b749cbc588ea28f9","revisions":{"dee4a1e255bae09a52bf11f6b749cbc588ea28f9":{"kind":"REWORK","_number":1,"created":"2021-03-15 08:57:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/780504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/780504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/780504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/780504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/780504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/780504/1"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 08:57:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 08:57:32.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 7.4.0","message":"update constraint for python-cinderclient to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I20a380a07655be4f5461c00a7033164a9d3bfc4a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I08740bafab73c7d5cd9a0f621cf8b25cd61bbe9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dee4a1e255bae09a52bf11f6b749cbc588ea28f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dee4a1e255bae09a52bf11f6b749cbc588ea28f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
