)]}'
{"id":"openstack%2Frequirements~780509","triplet_id":"openstack%2Frequirements~master~Ie84ddb11be46b31cbbe1384bb662ecbbd6f003bf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie84ddb11be46b31cbbe1384bb662ecbbd6f003bf","subject":"update constraint for monasca-statsd to new release 2.2.0","status":"MERGED","created":"2021-03-15 09:11:19.000000000","updated":"2021-03-15 18:32:44.000000000","submitted":"2021-03-15 18:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780509-new-release","meta_rev_id":"701f7faaec6c0f2962233cfee6dedb1692ed6bd7","_number":780509,"virtual_id_number":780509,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 18:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 09:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 09:11:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 09:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 11:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:15:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4118da5bbb869fe854acfa80acf546ac29023f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 09:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"951316d7e46ca8193ef93ae125ac128647a6a2d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 09:11:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c63f877bb004355464265d3331ad99cff8c4808","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:48:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/35379224b44e44aeb5bb5d450df085cd : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e437f95e2f78fdb8ca1e7008229676475c83da64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02ceaeab72047f680220e762041e051 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa34cb1e9c6436f8ea1ab9908ef17e0 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7bd742ed6bb41a2a2fb9a9e37005492 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f1afc2fce1144139fde2f578e528d8d : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4280ab3771540f9943fc04eabdc27cf : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b2be19c479346bb8b1e2f3b9db41bd7 : SUCCESS in 8m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50b0ccddf65a46f08439271b16d614c4 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/190d23791ff143289906e742ab2fe501 : SUCCESS in 10m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1267b854bd74a759c20ead08a00486c : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f618c79bfbae458b8135f06892306155 : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c9264860f47747d0bc15ad880948c75b : SUCCESS in 7m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83dd2e8c987241d88fc41d03ecfbcc2d : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/064715ef3ee84c808c97b99981146b7f : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/88ee764a995a42b599a65c1e2c31c7d0 : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09bdda4d86654023b03042d3f0e133f4 : SUCCESS in 18m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b730d64f140543aab37ffa82ef549e56 : SUCCESS in 5m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7dad2119ffc44cb89c0e531e03ff1e0 : SUCCESS in 33m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff10c74f2a3444a5bca14b88435cdafa : SUCCESS in 14m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43917f69e9e44079875b02a37cf050c6 : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e018ddb50cf84b23b076702594fc7cb9 : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e96cbd4fd57744db852818e2e469177f : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe4cef0a13324c33b7d95a8ea019c327 : SUCCESS in 7m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/35ccb3a6867d473c9b15bb2ffa11e38f : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1661f605de104fec83d82c6b07e7b2d5 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebd2743cdabe4ed0b4333e881d501ef6 : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1ebf55960649a96093ed011408c1c77c1015569b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:15:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b6c8ad39629f73dccfb29848550430a7daf6748","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:16:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a95fc32c8be8d780ac8bd3a3d2e7450db2735035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:32:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245db92ccab849c8b62a336fed9496ab : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc0a7d09a52481cad5b0c20e349dbef : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/446cfea16d4d421a97118b8bf4ce0b0a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea2f54a02ee849fe98df19e45e08b4ee : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32afa9a97ef648a39d61607f4d7c9623 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b4b0a66b2164ae78c72db265d723789 : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8261d9c71fe0464ca6da8f81f8c9514d : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9b41f8c4cac4d22814e5fc4c8f8d0f6 : SUCCESS in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9884324b404648fe9ea61bd287c471fa : SUCCESS in 11m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b587c62082240709ddcf0416303ae94 : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9397fa1a590d4cb49d89a61792f747e5 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79bb151395d448dc9c0ac0b6a344f4bf : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9285ad1d5d4d4c61921e0f54f5de7c2a : SUCCESS in 12m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0deb4561128d461f92302aae731d17cf : SUCCESS in 18m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/73a40cebca0e45279385c916f9e111de : SUCCESS in 7m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b3c855cc676145cd9395fde53d75ab84 : SUCCESS in 37m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac20d288520b4a6eb137bef325d70395 : SUCCESS in 15m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e5b0d99d4be47d1896d4f4bcf41eecd : SUCCESS in 6m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5d43c9c413a4a13bab660037cec2e15 : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/61c7ae24373e4162af8467ffad961afe : SUCCESS in 7m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6b70c43c08543f38786b11acaf5936a : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ecc3d9de531460da0188b32b6f3d477 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49981eaf595a476e885765e408d4605b : SUCCESS in 59m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70604b68738f445d8cfca458d362b953 : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":1},{"id":"701f7faaec6c0f2962233cfee6dedb1692ed6bd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:32:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdee4786833d15e3628ce2b53304d06bdf60cfaf","revisions":{"fdee4786833d15e3628ce2b53304d06bdf60cfaf":{"kind":"REWORK","_number":1,"created":"2021-03-15 09:11:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/780509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/780509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/780509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/780509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/780509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/780509/1"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 09:11:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 09:11:17.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 2.2.0","message":"update constraint for monasca-statsd to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If3a0423fe9ed7856ac09f348120f53bf99ff70a2\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ie84ddb11be46b31cbbe1384bb662ecbbd6f003bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdee4786833d15e3628ce2b53304d06bdf60cfaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdee4786833d15e3628ce2b53304d06bdf60cfaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
