)]}'
{"id":"openstack%2Frequirements~780510","triplet_id":"openstack%2Frequirements~master~I1a81d194c092fc6cd684cb3f54e2813bcc5de13c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1a81d194c092fc6cd684cb3f54e2813bcc5de13c","subject":"update constraint for python-solumclient to new release 3.3.0","status":"MERGED","created":"2021-03-15 09:11:32.000000000","updated":"2021-03-15 18:34:30.000000000","submitted":"2021-03-15 18:32:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780510-new-release","meta_rev_id":"eced703a6820fc8ee0a93ca62c71cad6656d6496","_number":780510,"virtual_id_number":780510,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 18:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 09:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 09:12:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 11:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 11:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:15:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f5aa3ac80dc1901d06fdc93b61870533057ec15b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 09:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5558640a513205a9df2c6ef159223d9a392e25f7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 09:12:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e32e7f708937cf51fc91bbc05adcbce24cfb2e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:33:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"210e655d021d97dc4f9ef9fa243434e2179df20a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:36:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c8e8c2166648a99b78602d5146ec77 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89621d5125943a2aa4473b647b13d30 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5e663a217534793909129ff49968225 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/879a9ca3416c41e79c531948d60926b3 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e05c6451e941738e46e993a63a9e05 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c33e3801246340a8b0fc9f5d2dfcd0c4 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9eccaf098f7f4c9d8a2699c69fe50989 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/272f3b5888304645839cdac52f95ebee : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dffa32bb60ab4dd09b3fdde34977c055 : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0cdb3319fece4a2dab8755e8c42818bd : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c91c087211f34764813569d9c0c5b747 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff2fac0d87bd41ac8310ab87625e9a29 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1683a6038bc41239dcbb0afbec16bcd : SUCCESS in 11m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/161b84d8d12140e0ac0b2d6650f01874 : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/37d3efa22f14446ca346138be40d5c53 : SUCCESS in 13m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3af69d69e45e4e9d9ef8329c9c6a78b4 : SUCCESS in 6m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f36ca13507d247d58a7b3f88dad090a7 : SUCCESS in 40m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c3650f6c795d40f18f6d1e934d69c53e : SUCCESS in 18m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/831abce0dfa645b98f26a4a8ae5124ab : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6483cb147ee140c5b8af6cfd3e534a3e : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/64b57ba21c7645238e822f99f4ae23fd : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c752d168ffc147459219f7bfd16cbb20 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4915ced920a4eb285ef78be7a09e380 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3703d81ec3c4d3b86440d3c5b10b322 : SUCCESS in 1h 21m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d0bef0ea6c5413d82f9f689333ace63 : SUCCESS in 1h 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e19d6556afb5e3040de6e36d2bcac76e10be49b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:15:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df728a428d246c2fbdde7842d2e9009d2409081a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:16:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44d58d12d5e7b595140fb4a8c3d5d3b2383541a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:32:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2aa67e598da4e3a8131ff43a02abc31 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8e0bb1c0304549bc0d6535f4f95470 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e189a094a44a4c7587a003facba197a8 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b48a0c1aa7f45d49314250678e9cca8 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cb41daaaaf145e28d896e5117af3473 : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f105959ebcb4ae29771cf75860a214b : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1177fdf28c7b4d88a1d3cabf9789a39c : SUCCESS in 13m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a0eccc162c2f43038a9419be969e6654 : SUCCESS in 12m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/370be6b48a054782a3ef972db1c7d4f1 : SUCCESS in 14m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dfa2175421394cf6ab9e137e3ea6f248 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c665ebeaddaf42b0bdd657b0282295f0 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb2698f3c76448b3ae9c2e602a547eec : SUCCESS in 11m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49de2b5623cf41418aa0a712271967d5 : SUCCESS in 12m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7904295aff5f4f1b807ebf941f562e0d : SUCCESS in 15m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b38bae8c88584935846e5d55ec6cd1d9 : SUCCESS in 8m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4cc8886e2cb94cbab64ccd51f6749ade : SUCCESS in 32m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b66e1afb2aaf4c90b96ca4b0e93c18f9 : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27423bfb74a04b40a5860f15c0d57c5f : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c266833e7d34687b1053964f877f513 : SUCCESS in 20m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d245f5bcc72a491198a52b15febe400c : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57b81a2a5ec94456989c6f804d31d858 : SUCCESS in 8m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0ded0bb328ad4bf4ae1fc646fac4ea7d : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dd39577cf8a49f29644a3b79e731aa6 : SUCCESS in 58m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/482d7311de61465080e14f94fb9f10e9 : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"96dbd5dbf4db9ca18719a2e9d9f9eec557db6af5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:32:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eced703a6820fc8ee0a93ca62c71cad6656d6496","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:34:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12cab3bfd904268ac74183d5ab04347 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fca38b0ce72f067c0d86b2ecb1df83d3b420fff1","revisions":{"fca38b0ce72f067c0d86b2ecb1df83d3b420fff1":{"kind":"REWORK","_number":1,"created":"2021-03-15 09:11:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/780510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/780510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/780510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/780510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/780510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/780510/1"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 09:11:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 09:11:29.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.3.0","message":"update constraint for python-solumclient to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: Ia2a0d3f1e10d1429a1bd68533128cd5be41b1570\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I1a81d194c092fc6cd684cb3f54e2813bcc5de13c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fca38b0ce72f067c0d86b2ecb1df83d3b420fff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fca38b0ce72f067c0d86b2ecb1df83d3b420fff1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
