)]}'
{"id":"openstack%2Frequirements~780540","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Idcd32edc49ef35ffb1027502dcba69c1a60c8ab9","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Idcd32edc49ef35ffb1027502dcba69c1a60c8ab9","subject":"update constraint for oslo.messaging to new release 12.1.6","status":"MERGED","created":"2021-03-15 10:42:55.000000000","updated":"2021-03-15 19:21:00.000000000","submitted":"2021-03-15 19:21:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780540-new-release","meta_rev_id":"6b2e3590e870eb2a0f00b17978052ed257e8ad76","_number":780540,"virtual_id_number":780540,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 19:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 12:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 12:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 12:28:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 14:17:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"64661b94b3735911b40852365039362c6c3ebc46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 10:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec2526a2a1efdf6c3ac70cfbbe5a061c2ab510b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151122638139407999fcf5a167f44960 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b38343d100c4bdea3b53b69b7d01cb0 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd0f5a5711da424c908f38429af3d851 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b44ba49445b14271b6783305a9153495 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49f6f0463151411c86a06522a8c36584 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b1bf2f71d884c76986eab03d41c54f0 : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/02cea9c064b8491b9a2b0406278aba02 : FAILURE in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/977d126f96fa480c8b0ab608addcbdeb : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c16aeb1e7124772b4c32f6e0c3eb049 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/358d82659f5344988ea33e13ed160797 : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5cad9d9049984254912d2920cefc889f : SUCCESS in 19m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7431d63d99a449dcacee28e88700969f : SUCCESS in 26m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c4f171fd77b412e94451893d94ffaa2 : SUCCESS in 16m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d031a93dc890448ea0edbdfc57105968 : SUCCESS in 14m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca7a0dc5688e4f86a901e40a510b9087 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa0ce1229d2044daa5564871f8fd5372 : SUCCESS in 6m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/84c68e29d7ab4009a64bf6dfa6d00eeb : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7801c78f57547f8b44f9ca478fa6824 : SUCCESS in 7m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a210489a53f646df80f7672979f3d8e2 : SUCCESS in 12m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b20b2e85e704549ba62f8594832224d : SUCCESS in 16m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a8c088d265445d7851f7a73391e49f2 : SUCCESS in 52m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11ae601ad74f4f5791f5d2acde74b75a : SUCCESS in 1h 13m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e255684ddb74438bbf9168b5b217ae8 : SUCCESS in 1h 20m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/080fc381ad774ee3aa584be4db802033 : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a52138707d4e429ce066577d91028096ff1b3d56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 12:28:18.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0fd37f8495e1f378d78f2223b16a5ce15279d95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 13:47:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718e30092edd4699967e507cf9d4a794 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67a3e2318d574e40ac7b93ae77e75fe9 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e0e28ef184747b7b03e2e0d8bb128b4 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b668e7c37044f698eef0143cf1ca490 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b7bffe8db994b58bb2befacbc9ce327 : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59f251d71c14440f8d6c75374581195c : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/850043db6e71405aabf966b58e29fc7d : SUCCESS in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f4736c95b58640abbf804aa4daf3d443 : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df24457afbdb46609c45a70fde89aa9a : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2238d302c6e74f7ead5ced5944aa884a : SUCCESS in 8m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/767d51011f6a44e5b8b2a64dcd01adae : SUCCESS in 19m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77b05f3007034ae49d0f0cebc8a49b18 : SUCCESS in 24m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0bce015b5733439c963d9e52c7a74db1 : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d581c63f94a4ba1b46804c1c23c1b7c : SUCCESS in 17m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5d93d37471f4d0c9c0fb483bd735d2f : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/309fa37edaeb4d8bb1503652e78eef0e : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2121939316ea43c5933932463e26032d : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4479a1bf7eeb4c53945f82d25080c162 : SUCCESS in 7m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e6ef266872f84d5d9246ed73f422ee5c : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/727b31018b2c4a03ac087102e0bfd6b3 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c08a9678e4a4e778b8720669adeb05a : SUCCESS in 51m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e799744f4c634e398de537f9b79d06a9 : SUCCESS in 1h 16m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34a85cb2122c4a2c849bea65fd1f5abb : SUCCESS in 1h 16m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/8930256c90de44f9b44d67f7a4cd35db : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9461cf4d4138f25f1dca7962297010a816d49a2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:17:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfe58a318597ff857203a4b03e2ae1ec48509f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:18:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b09098042f961ae841641bb34ee51839eb20ed97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:20:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702a6e30a231400a8cb0a9498239b941 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8b93adc0976462ea7c3e332f0fc46d9 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f85d6989894471ca282bd2962ca0307 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6034d3b587b34b1e8fa69f9d770e1512 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0184de818f9430d9da949c57f7802fe : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f3938dae11374768b187432e89d54eda : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d8d72c599e3494f8a9d15ff9d6e9338 : SUCCESS in 12m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d7c0103a08104309802986e36b4791ca : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a39a2086a6674801a3d036c1e738d377 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e99b2a2621b44c9bfafd277e85bffb0 : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/794c7264fb164232b2e771012557eda4 : SUCCESS in 25m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0cf43471f1eb432bb363ac4688bae2fa : SUCCESS in 28m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fbc9396f3a664e1fa2570ca6493be249 : SUCCESS in 21m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5f20d7e5f2c4681bb9ef13cf92c7f67 : SUCCESS in 17m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/968988ea6f844a19b0f60fc52407188a : SUCCESS in 16m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f105fab4fc640599ac5f261c4a1475d : SUCCESS in 7m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/979ccb9a81a34dfa8612c51bfd6d7d26 : SUCCESS in 7m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e7f602a8c06479b8d1ed038c6a42728 : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4cb1f2ddfc674eec8cad7028447e8937 : SUCCESS in 18m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9781f77e20fe4064b2e3925c26ad6e85 : SUCCESS in 22m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08afe4d835694e01a9cda5aff5011e03 : SUCCESS in 1h 10m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fe37cc9a9194b58bda5bb7b1ec824c4 : SUCCESS in 1h 33m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d6cf255777d424fb2e8bc19aec157d3 : SUCCESS in 1h 32m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/590fdea692c14c5dba67cc887e0b8ce3 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6b2e3590e870eb2a0f00b17978052ed257e8ad76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:21:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6acc69ecfe489272861adffd0c9e37b471fa416","revisions":{"c6acc69ecfe489272861adffd0c9e37b471fa416":{"kind":"REWORK","_number":1,"created":"2021-03-15 10:42:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/780540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/780540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/780540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/780540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/780540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/780540/1"}}},"commit":{"parents":[{"commit":"51a5c843a6198e1dd02423d5694ffd91bea68032","subject":"update constraint for ironic-lib to new release 4.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51a5c843a6198e1dd02423d5694ffd91bea68032"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:42:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:42:50.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.1.6","message":"update constraint for oslo.messaging to new release 12.1.6\n\nmeta: version: 12.1.6\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I6d006b42b6b4964e7b15116314934efffaa38358\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idcd32edc49ef35ffb1027502dcba69c1a60c8ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6acc69ecfe489272861adffd0c9e37b471fa416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6acc69ecfe489272861adffd0c9e37b471fa416"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
