)]}'
{"id":"openstack%2Frequirements~780541","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ie69ecb6fb63581a1d13cfbc066c1bb57c104df5d","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ie69ecb6fb63581a1d13cfbc066c1bb57c104df5d","subject":"update constraint for oslo.messaging to new release 10.2.3","status":"MERGED","created":"2021-03-15 10:44:23.000000000","updated":"2021-03-15 19:29:21.000000000","submitted":"2021-03-15 19:29:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780541-new-release","meta_rev_id":"22b83fdb72377864731fefddddf371b836822a97","_number":780541,"virtual_id_number":780541,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 19:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 12:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:17:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b89fcd1ce14492ec58931efa07c83cfcd8d6549d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 10:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45dfd0f3d6db8dd5d32e39e2d4c195dd9bd352d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:30:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60781a6bdfe948ccadf6c8d7798ac402 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82b4cd1e34ce40d7a048c89db980cfef : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f131275a87c476bacecb7292d37b7d8 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b62c974743d643c48a4e78bdad935b20 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/377a6f8ca9f141f59327957b26058046 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77ff5992e6bb4299820306e036a7d660 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9ea54a74a4724f9cafd98e81abf27b17 : SUCCESS in 10m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b7df629961c460499563d9214b6d89f : SUCCESS in 10m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f20cdcc35e704eb58bc2dffc05a9e75f : SUCCESS in 11m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b7942af11dd46ee82beb041bdf9e174 : SUCCESS in 16m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6481db9a9cb3469b945ff203dcba6b2f : SUCCESS in 6m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/66227ef1c6b041c2aa73b398e1435624 : SUCCESS in 5m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bc644180643b4c16bd01055300bd083d : SUCCESS in 12m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b1857ce2459748bb9f86506ef10cb3f5 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75abd581fc3549caaf5cac1d21d517e3 : SUCCESS in 7m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dc9d0a3047c142f2bb49baccc46375f0 : SUCCESS in 13m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba5d4db0cba34a99b4cec74598a6d4ea : SUCCESS in 15m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8abf8805baa14fd28ca7364caaea4fc5 : SUCCESS in 26m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5f5e76b5e3e440e9e892b7458ca32d5 : SUCCESS in 24m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/231dcf3399784e4da93534e7f8532561 : SUCCESS in 13m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/adf1effe6e5149bbb0f5fa060088ae85 : SUCCESS in 15m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dd9645c6a8324ca29782e0f09daa961d : SUCCESS in 13m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a282981f28894a4ba33f91ea0fda69dd : SUCCESS in 22m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b95444d7ac0547dbb5e2ff76b01fbe0d : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf5542b09a4140b8a1084bd652bc6066 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef29d2ecbda547b7b2e6f6e1144b5185 : SUCCESS in 12m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/096560d4008f4d85b4dee8b1cf230fa3 : SUCCESS in 53m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/781b63c54a4046ec9947a05d98c72a10 : SUCCESS in 1h 40m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ea834059dac44edaf086541fb7ec999 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f0b08328df680ccedd5b6fd991786c80b3d00234","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d40f76eda6fb51685e368627f758efc1f6247378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:18:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"369482e4b6edb3f1b9e72e7ea05011f0c375941a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:29:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ceb5474e544549ab9770f92fa1e4e7 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9202e15e3cc448cb0e79b93f758be71 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9460c7e3f13a4e449889eab209594e58 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11baed84be4e4d75a47978b5ff29918e : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2887fd0f99fc4bd28abfbd7abf111aab : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62354345d4274d8f93cf2de4c4baf272 : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a87eee72e79e478787871f415febd334 : SUCCESS in 21m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29b47981f45047838d0b3de8b8344872 : SUCCESS in 23m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9207429430a945d582980fefaf0bb139 : SUCCESS in 11m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b258fee2802f4977bf762455be985ce8 : SUCCESS in 10m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/98f1c0c9188a496d9764fc438adb6368 : SUCCESS in 7m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe955083215e4da8af92a1ab6bc7140c : SUCCESS in 6m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c9dcd44ff52b423287b12c5078f3d4c9 : SUCCESS in 11m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e2473ac8a6844b16be4493bae3e1150b : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d00e9ebf8b804d4583e8e7fb77cbdf36 : SUCCESS in 11m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a5112b4353a246d8b4e2f490ea76f12a : SUCCESS in 18m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f0d9e7a38bf4bff89a9f8d092928d2c : SUCCESS in 25m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/eea5976827d14fefa0559afda7795808 : SUCCESS in 30m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d830689a665b4e25b8e959e2b689664d : SUCCESS in 33m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/af5a71b45b144ffe8852e5066271a5d2 : SUCCESS in 12m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a57f013bd72a4fc490185f16b214ccee : SUCCESS in 17m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c1b797b9dd74492a9d86cc9d36ca303 : SUCCESS in 16m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da38022e09764350880a31b4fde45364 : SUCCESS in 22m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/012adda8cc224c7fa0358b3d34a7373a : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2c2dd2543c1e49d281272f5a50db13e9 : SUCCESS in 12m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d8e81ccc00543ab8fa2e2a3e2c0f511 : SUCCESS in 11m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/49baf3264c434fd6909400097efb3d5b : SUCCESS in 1h 07m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8994b9f81734471196d2653f03c06b17 : SUCCESS in 1h 38m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/464aef4ad5954d4ba6214620c4ece42c : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"22b83fdb72377864731fefddddf371b836822a97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:29:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"941756d9d88b5868dc1160b308deb721e020823d","revisions":{"941756d9d88b5868dc1160b308deb721e020823d":{"kind":"REWORK","_number":1,"created":"2021-03-15 10:44:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/780541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/780541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/780541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/780541/1"}}},"commit":{"parents":[{"commit":"9f905d08484a2280bc6da7a93a2c92a8e22c7dd8","subject":"update constraint for python-senlinclient to new release 1.11.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f905d08484a2280bc6da7a93a2c92a8e22c7dd8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:44:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:44:20.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 10.2.3","message":"update constraint for oslo.messaging to new release 10.2.3\n\nmeta: version: 10.2.3\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I6d006b42b6b4964e7b15116314934efffaa38358\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie69ecb6fb63581a1d13cfbc066c1bb57c104df5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/941756d9d88b5868dc1160b308deb721e020823d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/941756d9d88b5868dc1160b308deb721e020823d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
