)]}'
{"id":"openstack%2Frequirements~780544","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I053e156effc4f3c1895b8c94f58c20630229fbaa","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I053e156effc4f3c1895b8c94f58c20630229fbaa","subject":"update constraint for os-collect-config to new release 12.0.1","status":"MERGED","created":"2021-03-15 10:47:28.000000000","updated":"2021-03-15 19:29:38.000000000","submitted":"2021-03-15 19:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780544-new-release","meta_rev_id":"cdcd2efdd24832db95dd180d9aad30a33b432901","_number":780544,"virtual_id_number":780544,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 19:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:18:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 11:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 12:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:18:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6c5890bbdb34f8f691ae2b6a33ce3750ce45f52e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 10:47:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc87c8498e328d91ee2ad454ce7f928f4eec684f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:08:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b20d8c44db864c81b3f3cb679e15fd64 : SUCCESS in 16m 13s","accounts_in_message":[],"_revision_number":1},{"id":"165552850b8f750c5fc3ad4602c74c9dc18c5ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:13:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3676c75f8b9d433690dad1bc4c59b6eb : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f9329d145ec4b689c27eb5a022b32d8 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73d469fff4074cf3975f410217580541 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46f18ba285d74601b39ed1804cb4cdc9 : SUCCESS in 7m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6609e9841e7944e68e6678848de072ee : SUCCESS in 7m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ac47bb9955540c4a6ee36094930142c : SUCCESS in 10m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0da4265950a439c9683ebb652b300cd : SUCCESS in 11m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a0ac8adc603b4aaf8d3f83705bd04255 : SUCCESS in 12m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16453481c16e4487906ae1ceb1a79a73 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/95f7382fc0b2439e8e6b14d42fe9b61c : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/856fb0fb48634b25920068c53dd08176 : SUCCESS in 12m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9d022121c64d4bcebc2225005c394d73 : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/370d266b107c44e69054feabee6932ff : SUCCESS in 20m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/037730f2180f4d958d8addb34175809f : SUCCESS in 26m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/96bb4343a0b14401a6a7acef59fe259f : SUCCESS in 15m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d246f2111ef84603875a3e523c7a373f : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a765b09101c43449e8ad6f0143c0a9d : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9da3e18ccfa4edca3e608f1bf2bc961 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9a22af7e3f84e0386e3b01ca121c2fa : SUCCESS in 7m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a9d8c8f103e499ba7f904dd465490dc : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f860b6913234a58a3ee113389885b04 : SUCCESS in 57m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af03a7652e1848c39b93822d79847080 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e100a037e168b13125d197b1d8b564ceb016bd1b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:18:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"723f567a74a84e679f6e8ab9a03259b29530478f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:18:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"777e23ee057b263b9f27218efbc2488bdecb2f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00dd8f5fb15241f491fac3f1f118444c : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4666598de80348c2a7234d9fa3fc9adc : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea6bbf32eb40400e8574bb12f1364f26 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03279e1476bb48248523189c164c0b1e : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ed6ab77cf9343e2b85f999a1458ff95 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12440b264aca4f429ce7d5a9444282d7 : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9dcc598edcac48f99860874120e6fdb4 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/675c320b755b462e85b91218496cc9b0 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b017043728224e55a3c650cd547e148d : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45ac7f6e850f4255960735e3fcd74ba7 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a214c0a4383345b4b06701b473efea73 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f3d85f37b924b718997be498eff862b : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f5ab2c5baf5d4201baf3b31bd3be31c8 : SUCCESS in 22m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64b86485bf3d41389f0e66ddc203f283 : SUCCESS in 28m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/717ea5cee8454584ad332fe89eeaabed : SUCCESS in 14m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54e8f831fb5e493d830bc5fbf823cded : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d200989736a64d369760589a1eb40d9d : SUCCESS in 17m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d7a266371594677b2f74a03e4c60a53 : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90eb7fa61aa749ed9a204433a5f7d326 : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c7db83f041d94c02990c5e05e59e5e29 : SUCCESS in 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69325cc5abf74feca54b0bd2bd83bbce : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b744d5f97184f88ad48463c6b79e904 : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cdcd2efdd24832db95dd180d9aad30a33b432901","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"433aa9ef398b38f62006719819d88bf4765af08d","revisions":{"433aa9ef398b38f62006719819d88bf4765af08d":{"kind":"REWORK","_number":1,"created":"2021-03-15 10:47:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/780544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/780544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/780544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/780544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/780544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/780544/1"}}},"commit":{"parents":[{"commit":"0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38","subject":"update constraint for ironic-lib to new release 4.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e97a6d7a11dce9bf2571f86de2bf3ba48daaa38"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:47:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:47:26.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 12.0.1","message":"update constraint for os-collect-config to new release 12.0.1\n\nmeta: version: 12.0.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Iec0aba1f6503e1b92e84f1b7150c3cf556c1a742\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I053e156effc4f3c1895b8c94f58c20630229fbaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/433aa9ef398b38f62006719819d88bf4765af08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/433aa9ef398b38f62006719819d88bf4765af08d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
