)]}'
{"id":"openstack%2Frequirements~780545","triplet_id":"openstack%2Frequirements~master~Iba90ad994772a40df6caa201f3775e0ca63bf3d7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iba90ad994772a40df6caa201f3775e0ca63bf3d7","subject":"update constraint for horizon to new release 19.1.0","status":"MERGED","created":"2021-03-15 10:48:35.000000000","updated":"2021-03-15 22:39:07.000000000","submitted":"2021-03-15 22:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780545-new-release","meta_rev_id":"47625ff11cb43eecdbf4fda13e46688d4238b5ec","_number":780545,"virtual_id_number":780545,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 22:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 14:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 20:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 14:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 11:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 12:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:15:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-15 20:37:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f7eebdc0e7fa6bd813f7aacef3c967c0893ec2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 10:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e897063aa2e784760cad58742e676c3bc0ead1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:25:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77146e35df844095a280e260b96a4aba : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"107d47aef17003ce15b906f8d53a75fa2ca9d3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:15:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e12b5eae4645cbb6279345ab58a1b2 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9523afdb98444eb974a6b3bcacbb4b : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aef37ea324554676a3447699cb6deac4 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f9345fd311045bdb0c0d9ba27980f50 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3b192d094b497ebb8898a3aa9b8a3e : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ad9781df7bc4ab08cd720b1c4519da6 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/338b6868effd4c9ea04345ee6710bcb9 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/757ea50a93c541e7836d12d8823cc51a : SUCCESS in 9m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/886561c863a747968866f825408b359b : SUCCESS in 13m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ddba031acc9454cb0adc46ec85bdd3f : SUCCESS in 13m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/84b029251b50419c8a7f3e707bbdcb30 : SUCCESS in 8m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13fcf7ad7a5846348f9d1bb6c21df68f : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b4f6e8b34764f41aa16bd1866f58d19 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3bfb9001f37e4c58a308956be0c700ba : SUCCESS in 10m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4293c21c81ba4ac39424e06ac5fa22fa : SUCCESS in 18m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f8abba40857749d7aea42fc421842e36 : SUCCESS in 8m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/87004763c29f46c8bebc57f98b5a3408 : SUCCESS in 38m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bbfd8f4b3b7d4ed4b80da3e456dd9a20 : SUCCESS in 13m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/465b848653514000940e5232fd117053 : SUCCESS in 6m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e07d7d33d0304de8972ffe52608ab56d : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/794689d24e84494497a81310a84bd575 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/444e497a0dec493ab327980da91a78b4 : SUCCESS in 8m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf22805220434af6a0a2f2b9204e36fb : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/745c7228cc9149728fe0b118c6fb0192 : SUCCESS in 59m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6ded6d319e948eba9cd6585fd51388b : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"66eec77e737e5451986585aa145f83f8c8e6e9a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 14:15:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abc514adc8272ed050936bac5eb0e8d19872889a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:17:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d5c46716ca19f76418aac715b2b2a9ec24b8162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:33:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367c4af9eec74ec08fab0c683cc01214 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2b8ac0eda44a8187320dbda55c6965 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a289ed1a39cc409ba3df781c93096e38 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88ba377dac8f4b8291409917c0197d26 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8d087d9892c4f82bad22c83492d74a7 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c699d451f394b2a978bf938d804c296 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c1de3e78ba3f4af7abd0bff4d83e7eea : SUCCESS in 13m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/325c3f496dd0480e9f13867d7ded3a5f : SUCCESS in 13m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/187a91d743fd407994d2b34f1e807165 : SUCCESS in 14m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3ac0b72723545d7add0f3ec2bbe76f6 : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26129b86de8447c1ac6defa327af8794 : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e056877840614326bcd22bb8cfca6bcb : SUCCESS in 13m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c8b980c6b40d4480bdd41bdf2134c92b : SUCCESS in 19m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d998d4c8119b439383bd2dc091b9521b : SUCCESS in 13m 29s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2add68c6e3784c68a09911029daa2fdf : SUCCESS in 8m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8e0981925164302bee602dbc70f86f7 : TIMED_OUT in 41m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39a7a64d524c40b6b7f20f24d6435e58 : SUCCESS in 22m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed56b00ae35c4495bfac994c7b001822 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d2f4a9e7ad44e288c9cbef71bbe47bf : SUCCESS in 21m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bd486e69304941c8ac8043bec848303f : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85eacd08fef44602a9216d6975109a6e : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9bfe89d49e4a4dcb8a30f2b83f560d00 : SUCCESS in 17m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4468c0930e146f5b361b839e1a78518 : SUCCESS in 1h 01m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98acdcf152404766975620fabc48ac2e : SUCCESS in 1h 26m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3c54c03ef1b23d558c672fb40c84815c92bc09e6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 18:52:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ecc2273a931b1779c699e735cb07523befdd9a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:24:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/528ea923fd96468b8371a859f16c4f8f : FAILURE in 28m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2791eb92a644bd146ef76aa5d5483b69463e28fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04808a9a4a445fd883564e78d491b0e : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34bb6d4764f046a1bb244d0992d60441 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b11994097444635bad1ff0a82d6d818 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a74d782a150f46b48a8bb761d69d354e : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b42482c914e4b44ae2bcf823964a32f : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08d0ed84e35943779b8003a839253a8c : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/281c5eb41a7f41e6aeaf5dd434b7939f : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d091b1a30b974edc899b3d101b813074 : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/717bbe47bce74795bfd0a9bb21541fc7 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d74d9f0a93254694874d2bd6d295d635 : SUCCESS in 8m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41aeee7ee0e047ab919f92e24cf2cc3c : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1b7579f46814133bb3b05f413abaee7 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f49ea6451ec944809bd4fe7bd7e106a8 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/62b6ca0e9bf6492b86609b265cc6a2c3 : SUCCESS in 11m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0c04795e5584b5fa2576169965f44e6 : SUCCESS in 12m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/bb660e2c481647bd99c7ab11942f0643 : SUCCESS in 5m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/517021d6c28f44458b6e5fcc1a3935e9 : SUCCESS in 38m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84d22894e3e64f15ba643154731f07e9 : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3493432f26ca4961b902aa6f2c62331d : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc5752a9519d4f6e87607fd11544ccef : SUCCESS in 19m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65066f55e5174cbe9770e97c97d03a4a : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a0c4854b33b46b9bf33bd351dfbf8d7 : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5030de7142cb48ef91305868288ea7fc : SUCCESS in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25a7599af9d84187b0a46c583333eb1a : SUCCESS in 54m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c32d7aac9851442b94ef8625ce87a4d2 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cf90641ceb425ded615edfea46c430597650403b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:21:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cb1fd76355df16b8b5077e781660c3dd13244d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:37:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62ab4bfcc491e25647d00490ab4db0e1fa4c6c66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:37:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5434eb600b348ec893f21f0a65fb2f4 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904f2350018746f3b9d261577a4889b4 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49d4fbfb8f724f5d96858944958d1d61 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/141119b357334a109c429500904c2e20 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89d21c6a015c4d059eb7237782b3238a : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d2ec5a68d4e422994e48db5c48e9792 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9b7d6e65d32410c894b9a0b46e7719d : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0bfad8b6d6ae4a4aacbb69cf43fb7768 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89a8e1c6d4694dc78bb6a89d40016fd6 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57c46becfc4543d3bca5c2cc91ca6991 : SUCCESS in 8m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9035746db70485196ab12befe2a5b77 : SUCCESS in 13m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cbf70588c8647a2924e795409c18307 : SUCCESS in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0f013c55cb847d49813af30412b1eb4 : SUCCESS in 14m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/efc27202b0d6468181eda1ac3f276c3a : SUCCESS in 13m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/03bd6244c86c449890c4a0a3f98ac3d9 : SUCCESS in 7m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63408339a85843179b5e706de66c34e3 : SUCCESS in 30m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6baf267e06a34631bc1697bd7e0ab190 : SUCCESS in 10m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68b04b64348f48c1a94e7f4908447448 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee2e6e110d5e451f9e920509024657e7 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9f46725a847b4da98a77ccd6805306cd : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/482b48411ec040a898e749703f6e9ccd : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ca9ae1226ba54fd8b2410a6d45d49c96 : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a22c6be583994b278c1b8d9d6b0db28e : SUCCESS in 37m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa58c4b42f9e454aa82a3e9010bde6dc : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a1bb7536701d1b1f480304eae614c728ec93ca92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:37:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"47625ff11cb43eecdbf4fda13e46688d4238b5ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:39:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52247fc0d334640ae0642a6b903b53b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f93458302b0fca4e61087168ed55455534155121","revisions":{"f93458302b0fca4e61087168ed55455534155121":{"kind":"REWORK","_number":1,"created":"2021-03-15 10:48:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/780545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/780545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/780545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/780545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/780545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/780545/1"}}},"commit":{"parents":[{"commit":"8964363e4dd2374cfb4b1d2dc1d699dda7a7f964","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8964363e4dd2374cfb4b1d2dc1d699dda7a7f964"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:48:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 10:48:33.000000000","tz":0},"subject":"update constraint for horizon to new release 19.1.0","message":"update constraint for horizon to new release 19.1.0\n\nmeta: version: 19.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta: release:Commit: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta: release:Change-Id: Id0a59781b81aed9b99e9495e60554305b82e0b7a\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iba90ad994772a40df6caa201f3775e0ca63bf3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f93458302b0fca4e61087168ed55455534155121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f93458302b0fca4e61087168ed55455534155121"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
