)]}'
{"id":"openstack%2Frequirements~780667","triplet_id":"openstack%2Frequirements~master~Iff9b2f99ff75791c67a345e10dc65f29d76e0323","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iff9b2f99ff75791c67a345e10dc65f29d76e0323","subject":"update constraint for python-manilaclient to new release 2.5.0","status":"MERGED","created":"2021-03-15 20:44:43.000000000","updated":"2021-03-16 00:05:10.000000000","submitted":"2021-03-16 00:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780667-new-release","meta_rev_id":"c704a747bf0c2ba9099eac87dcb19b91815539cd","_number":780667,"virtual_id_number":780667,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-16 00:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-15 20:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-15 20:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-15 20:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 20:56:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 20:56:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-15 21:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 21:57:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d2f9945319a19adf4d3b071b70fb211b722dc60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 20:44:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b2fb135b99de68bbd665adbd352097eb841d69e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:56:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43ce4b5c2ea5998e24f2f5e5870355541f0b3653","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-15 20:56:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8a970de2da47ef8f2f9b6a7ffaa62a2626b4015","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:57:20.000000000","message":"Patch Set 1:\n\nthat was the last missing patch for libs freeze","accounts_in_message":[],"_revision_number":1},{"id":"777b21464994397111f88a810e81955536d43152","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 21:24:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d712e480c294e71ba0d2541dea0c3f9 : FAILURE in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"06904687d69d6bbf1cc769ee1d08ec9b8e4c918f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 21:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153a9ae25dfc426c887b9a393902fafb : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5c70c4a09a48e0b07072ec9cb517a4 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d4665bbb14843be8968e80afce65327 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33ffcbab32854801bdcce6984981b5b3 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d05c464abb9d4e5d9f9dacf9989eeb86 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bde53c1624446a586309c797522ef8b : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5419b26d7d184d3aa8c01be20905e729 : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/377e3c3cb97a41d89d23df96650e81cc : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56e1651ab65a4eb68b2e121f6af03ce2 : SUCCESS in 11m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/21589307e07b4924a1fe255bb88e2164 : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/691812a3a3bd4c54b251fe4d3b7d8bf1 : SUCCESS in 7m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ccebc44674c4e829c1b2d15ecf89ff3 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/457a4f38a8104966821b40b9af26b60c : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6dbba15135a4b15a95c005e9bebf52c : SUCCESS in 10m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce8fa528fe094720a5d3ca1929fb28a6 : SUCCESS in 18m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c9a30236f6744eba8e4643a2f87ac7df : SUCCESS in 4m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bb1e5434e7454a9895170781b7e84edb : SUCCESS in 34m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6a44ba95b74d477798efb94db9f24bb7 : SUCCESS in 11m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95bf8d8f1f4e4d478b608593a42d03d0 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/873bd0be754147eca0217701b3383872 : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9ee7568bb1ef40c49949f9f7f4a12ace : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b1ccb4764384dc19342e91760dbd435 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/819401b6a5984444bd5ad7f79b1632fe : SUCCESS in 16m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c91869f94f314bd99bde7867023cfc1f : SUCCESS in 1h 03m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4091081ef5684523aa20310e1e003329 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4693cf4c6d0ae66c6d122171686946c262266e1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 21:57:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f560895e7b76f5f641ec45b470ce97bfd668b618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 00:02:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabdb84291364ca0b8b6051e5a080394 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbabb0f2d6a4d1e8eff34a24ad1eb17 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e0704a15aae425dba248bfe56c3ddef : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b594f475c3a416cbe5109cf7c7e8b33 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea563be5b9fd4585bdb1ed7618eda952 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5cf2743a989b4900912630ca20c9b48a : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5259df4b36484a9da36c703cfc2bfb05 : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95ca4e4aae324e5f92a5b057a08d3e0d : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d50c73ddc9d0464b93689ea85e8d9fe0 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5e23bd5239247d9b8bf515977d57003 : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee246e9ae4c9458393103057b5ccdf16 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64047b7b711944c98719f306ce8f1b60 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/87765a87c9e94657910fdccdee107afc : SUCCESS in 11m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af34119d77ac4cc7af4692c41e6ed5d5 : SUCCESS in 14m 19s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/91d6338d8570426b9ee9fe406c731f51 : SUCCESS in 5m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/79e09826962744b69cbb0d2cfd532cbb : SUCCESS in 23m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9738913ee9df49fbada806dc376eea72 : SUCCESS in 18m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/919298db1c0e4be68d8e425eb929122f : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6ad4235616d4fe58ac8fd7ec34b3ab7 : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1850b4a5b3894ad2a05e66806543a3c8 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c538880037cc48949f467ede513cd595 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/53c12b1d76db438c845d865bfe90a09a : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8282a863e90f489c821a32ef034df113 : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fccca36e11640a098fc05d187a5166e : SUCCESS in 1h 51m 44s","accounts_in_message":[],"_revision_number":1},{"id":"76c9c2d28d59cfdff6473d04fe14929b097beccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 00:02:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c704a747bf0c2ba9099eac87dcb19b91815539cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 00:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca5e7f3e80848af802799ccac67afc0 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcd931ad3f41646cee486e50c14a472d16469719","revisions":{"bcd931ad3f41646cee486e50c14a472d16469719":{"kind":"REWORK","_number":1,"created":"2021-03-15 20:44:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/780667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/780667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/780667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/780667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/780667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/780667/1"}}},"commit":{"parents":[{"commit":"8844a13094a5ed259e2e7073cff208febb72b43d","subject":"Merge \"update constraint for os-collect-config to new release 13.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8844a13094a5ed259e2e7073cff208febb72b43d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 20:44:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-15 20:44:41.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.5.0","message":"update constraint for python-manilaclient to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I55b40d00a58128a652560aba86d0be5712712ad9\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iff9b2f99ff75791c67a345e10dc65f29d76e0323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcd931ad3f41646cee486e50c14a472d16469719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcd931ad3f41646cee486e50c14a472d16469719"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
