)]}'
{"id":"openstack%2Frequirements~780860","triplet_id":"openstack%2Frequirements~master~I6c8659e396f74dcbe4d3ad874674b1a38e46df4d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6c8659e396f74dcbe4d3ad874674b1a38e46df4d","subject":"update constraint for neutron-lib to new release 2.10.0","status":"MERGED","created":"2021-03-16 14:21:58.000000000","updated":"2021-03-17 00:41:31.000000000","submitted":"2021-03-17 00:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780860-new-release","meta_rev_id":"3ecf77e63064672fcc34d3eec3212a160e3f0abb","_number":780860,"virtual_id_number":780860,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 00:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-16 22:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-16 22:27:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 14:43:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-16 15:00:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 15:39:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-16 15:40:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-16 15:40:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-16 15:59:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-16 16:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7677c2f850d837a4fb02faf20d2091b6ad4b4e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-16 14:21:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"801721d3aa22c747816b3799c1678ef20e9c29c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 14:43:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ecb53183f498cd94671e5c07909317f85fb6a1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:00:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/38c194f01abd493f9de83073be0944eb : FAILURE in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b1e41abba1920b537d2aa1749e78516158f32a31","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 15:39:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17397144ec73ba36051a7d61d03768b11a74e6bf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 15:40:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"376493f83e5c98831c04757b8f318a069320385a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 15:40:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b837e4d0928f60c0a73d7739ad2ed4d0ac4ff20e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-16 15:59:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d32473d29c6911ef4d6cf058b4fce88c1e083c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:54:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f96c77ba28f4ce2bf620462eaec9588 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af946eef72024ff9bc63e9a2ac3349c4 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7e2f96c7284324bcdf8c2f467b99dc : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5585019ed4e4a03b12c6dfa7459dd82 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ed63c7c9114f10ae4ece62489bc431 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6aa61bb485f4b21860c6a846b060a63 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25b956732cc34e069bfa824cf68071ae : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b27f086e3a4540d8b3612eb9261de920 : SUCCESS in 10m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef1d1cbc058a4dbf8bff54a7ec8c6e49 : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cbc7eb7078f34616955c92742723bb03 : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce9bfe48d62d46e79af22892d630d9b5 : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/565448ef4c58495a86bbd64cd1087da1 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e51b93bd1b34f819bf9638a6e56fe26 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/421ed4f4017b4f3bb90fdefa7040e349 : SUCCESS in 11m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8588cbaf4ff34e138773f57fac628866 : SUCCESS in 15m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0e838cd604304d608cb5373c8f378ca8 : SUCCESS in 5m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1cd28939f3d42f887773a02243c7d81 : FAILURE in 30m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7340dd203375451b835d93989e4f2d99 : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d3e6306d71d480d8529741fd42b49d1 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/003714923de24a5382d3b868112b914e : SUCCESS in 19m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/acf3d841b33641d2a206c71402c60dc5 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a97b4bfbc75f46949e9677825c9f7a6b : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/faa539953ea04fe8b536dd65a3c7dbb5 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eda8b06957441f6a78658e2832cbbdf : SUCCESS in 1h 14m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/100078d693a847848f721f6d145659d0 : SUCCESS in 1h 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"783efdee9bf545c29847c727b4bdebaf17ffc6ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-16 17:14:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fcddbd9bf4f1730bc64a05e235d89e61ff6f3bac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 17:19:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"17dd1a37f9118c95e02e4cd6ec0e37b86190626e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 18:44:51.000000000","message":"Patch Set 1: -Code-Review\n\nThis is not going to work until [1] is merged. But [1] is expecting this patch to install neutron-lib 2.10.0.\n\nWe should first disable the Neutron tests, merge this patch and push [1] again enabling those tests.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/780802","accounts_in_message":[],"_revision_number":1},{"id":"dc878ef0a8d1fc34d46e6e3f39c085c6ded41626","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 18:46:04.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e This is not going to work until [1] is merged. But [1] is expecting this patch to install neutron-lib 2.10.0.\n\u003e \n\u003e We should first disable the Neutron tests, merge this patch and push [1] again enabling those tests.\n\u003e \n\u003e [1]https://review.opendev.org/c/openstack/neutron/+/780802\n\nBTW, I\u0027m talking about these tests: https://ad2781ccb869c0ef3e99-0fce708e89812c28d28e2be1213f1c9b.ssl.cf5.rackcdn.com/780860/1/check/cross-neutron-py36/d1cd289/testr_results.html","accounts_in_message":[],"_revision_number":1},{"id":"747727d690b4deb9daa3c2b9bdc5ea095ffc0625","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-16 18:51:17.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: -Code-Review\n\u003e \u003e \n\u003e \u003e This is not going to work until [1] is merged. But [1] is expecting this patch to install neutron-lib 2.10.0.\n\u003e \u003e \n\u003e \u003e We should first disable the Neutron tests, merge this patch and push [1] again enabling those tests.\n\u003e \u003e \n\u003e \u003e [1]https://review.opendev.org/c/openstack/neutron/+/780802\n\u003e \n\u003e BTW, I\u0027m talking about these tests: https://ad2781ccb869c0ef3e99-0fce708e89812c28d28e2be1213f1c9b.ssl.cf5.rackcdn.com/780860/1/check/cross-neutron-py36/d1cd289/testr_results.html\n\nhttps://review.opendev.org/c/openstack/neutron/+/780948","accounts_in_message":[],"_revision_number":1},{"id":"c08bb59fef01cb7cc29204e3c7921920f310c774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 19:18:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f7144f2ae54c43b9dd462b430a83ab : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683ce054d49346f99f568581c4182f72 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a0953dbaa0742b9be75a5aba06c71c6 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86eac9a2a57841058b21bcefa4b80957 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a634d7fe64e494188a73034da541480 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83d302ed390b48728a571c3bd150e1ad : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5daa141877a047d393f3462ad86895fe : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a215e6b2787f4ac78db757cc32244db7 : SUCCESS in 12m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30b260be52274f1caa84440167a6e02c : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/792eb7df49824528aad1ee035a24bee5 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6de488f66f54419686503f001f16e4c8 : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9f2aa99a3b1744ae87b767a8ef66fef0 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fccc19f310a439e9894b5654e0bc598 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2083533b287414a8f463b14ec5ff49f : SUCCESS in 11m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80cab9f1491444b4acb6dfd09ccd2e00 : SUCCESS in 13m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/00baeeb3d2ba4bb49af9e0c8aace23e7 : SUCCESS in 6m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c20a6504df4497c88aea52b4a3f344f : TIMED_OUT in 42m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef9f78ac82ab477e9e752b41b44365f3 : SUCCESS in 16m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80e05c75ac354757ace1597baac98ae1 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75d86909c72d46469be81538e5e45945 : SUCCESS in 16m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23b42a197b334537bd743298891d6b99 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdfa8881df4c48be8bf3ef3a3bf14d5b : SUCCESS in 11m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a1c779cbd3d84b5dafcb215fd7a5d509 : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/404d46f368664acc89e8c3da587e881c : SUCCESS in 58m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3cef07cae6446b9aed6b4a46ce0f50a : SUCCESS in 1h 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c04dd57dabb4ec951874e7195a5d14a7e6241720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-16 19:59:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"685a5c069877e06a6d4f82385ebd64765db756f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 20:39:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d951cf521f474df39f8cae9e753a16aa : FAILURE in 34m 44s","accounts_in_message":[],"_revision_number":2},{"id":"730ee800c5ac57e4579ff5b5712ae1016ea7db00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 21:32:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e0a682e03d42cbbea9c0d00f178fc4 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e800f376f0427fbf06cffacf32e463 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f678a9a7f943d8b75624f6d50cc57a : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2af43e6b49d4ac3ad162d25afae7d4f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd7f4581761648078ca9e1a66ec4629f : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1286a4351e8450a9ac0bee2a5c8a416 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55eee74eeb6f4f928a1bd20ad7c230a8 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdc211f92335498ea6365572a306dec0 : SUCCESS in 17m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1165e7e23938443fb211bbd0730a0e30 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/590f8193fe7c4863b48b833a1cae386f : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/834453e0c83b492988d33ccf712bfd44 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a1268c420974e05a801017c49dc4d5c : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7cb75eb6d1549489a386ee5bcaf676b : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6504102b8c2649df9d8822875f7a018f : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/84402d79b69a456bacc4f94cd46db8cf : SUCCESS in 14m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/cc3cc5a8348844009aa693dc12fa1679 : SUCCESS in 5m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36197235c2614a029810bdb576c9980a : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/693f74967e8a479cb85888249de27ea0 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74481e30aa184c2bb266f71472058ce3 : SUCCESS in 21m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/36503ac9d48f4b4ca4971cf66196ffd8 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb2753794bf94e3c922718b495b9cf24 : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fee62d19dc134ceabff0a182e9581f7f : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7561e702e65040e181685d70326fb088 : SUCCESS in 1h 03m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a7f8376f647493fbda1e2d9135ec12e : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d380446a770f6c4848d9c962686933d640271948","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-16 22:27:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"028e5bd7dbf3963525512a0fce9da1117f08ca54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 22:27:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"133aa98b21ff7bcf382df08a23317bd2a1026788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 00:37:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79fb11a083f4cd6b57011f111b9b05a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f511cd1d054f1a8acdd9710cbc2a16 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda9c2971ecd4d4a83ce8a69b18d10dc : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1824e93abf94b9b8beec322f4dec8cc : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09b6f75353be4862958fd71964165a7d : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc3ac99204604306b174e0e0d5b53b51 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/968cf0df247841e297222d3e42a89a9f : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80a856ce319d434f97171b5fd61d5116 : SUCCESS in 12m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e131d688f06e4168895462947c537a54 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/833737a9d4f24e18b652aa47ee4cdc62 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b8c4beb430e4dd997bbe5a014e2ae77 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e37c81fcdfc94d9d8d4184630b407203 : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/377dc935b62a466495bbbcce37fbd7af : SUCCESS in 9m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a673bbd2464645ebab60cc02e2c2376c : SUCCESS in 21m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/bf0aae411ce64a70ab44623927dff5f9 : SUCCESS in 6m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ba4ac5516d784c4687250860572890ed : SUCCESS in 14m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/157e6bca56f842fdb71cd76b85bd4bdc : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f4acd8fdbd24151a50c61f31332dfe7 : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d7dddb942b4640b8ac1cc4d2072f0b7e : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c586f31f67347d08e177a1e1b6ec572 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c72b11270654dbebbc1d85de1f02dd1 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bd2be1bb41c46108b3b09e00e4b0e5b : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/263b2b2b58b6464681c7b9ca62bec3e6 : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":2},{"id":"be38394730229b1780d896f1c769b7dbee563d03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 00:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ecf77e63064672fcc34d3eec3212a160e3f0abb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 00:41:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdf18b0ec124f9f92084bba6814ce03 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29eac2463669cd7ea48c640715f7c518991d83aa","revisions":{"ae0affe90a1cfe7652c720db811b138f6bbf49c5":{"kind":"REWORK","_number":1,"created":"2021-03-16 14:21:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/780860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/780860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/780860/1"}}},"commit":{"parents":[{"commit":"c4279a3522ef251b0426dd0a0a5dbc70ec009081","subject":"Merge \"update constraint for python-barbicanclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4279a3522ef251b0426dd0a0a5dbc70ec009081"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-16 14:21:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-16 14:21:52.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.10.0","message":"update constraint for neutron-lib to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I6d1cbc54f36ee34b51420b1c2ed534b0eea53afb\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6c8659e396f74dcbe4d3ad874674b1a38e46df4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae0affe90a1cfe7652c720db811b138f6bbf49c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae0affe90a1cfe7652c720db811b138f6bbf49c5"}]},"branch":"refs/heads/master"},"29eac2463669cd7ea48c640715f7c518991d83aa":{"kind":"REWORK","_number":2,"created":"2021-03-16 19:59:44.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/780860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/780860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/780860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/780860/2"}}},"commit":{"parents":[{"commit":"c4279a3522ef251b0426dd0a0a5dbc70ec009081","subject":"Merge \"update constraint for python-barbicanclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4279a3522ef251b0426dd0a0a5dbc70ec009081"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-16 14:21:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-16 19:59:07.000000000","tz":-300},"subject":"update constraint for neutron-lib to new release 2.10.0","message":"update constraint for neutron-lib to new release 2.10.0\n\ndisable the neutron cross test as suggested by ralonsoh@redhat.com\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I6d1cbc54f36ee34b51420b1c2ed534b0eea53afb\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6c8659e396f74dcbe4d3ad874674b1a38e46df4d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29eac2463669cd7ea48c640715f7c518991d83aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29eac2463669cd7ea48c640715f7c518991d83aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
