)]}'
{"id":"openstack%2Frequirements~782136","triplet_id":"openstack%2Frequirements~master~I297916c3b4a4c6756350b21868f1d57c26937208","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I297916c3b4a4c6756350b21868f1d57c26937208","subject":"update constraint for ceilometer to new release 16.0.0.0rc1","status":"MERGED","created":"2021-03-22 09:49:22.000000000","updated":"2021-03-29 19:08:12.000000000","submitted":"2021-03-29 19:06:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782136-new-release","meta_rev_id":"e2a052786f96e401b2d78e592601cb3f2203116d","_number":782136,"virtual_id_number":782136,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 19:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-29 14:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-22 11:56:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-29 14:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 10:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 11:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 11:56:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-22 14:01:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-03-29 14:12:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"173ca29347799f4b2ccdb7f151a39eaf3d2adc21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-22 09:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb3102f1ab76c6bab8f2b1965ec085e0fd67f934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:37:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5511642e74ea477abd9b24ec390a44b8 : FAILURE in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8d5afcdb8b3dc8664f90f00e7667984738460ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 11:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a26aac88ee478e804e97460d7c62e0 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e319be43a474ce8b485aa2d2d8f1c27 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff6d8e8e638b4fbaa2a696c27226ceba : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2a80bc62def47609d2bc0cb17524770 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/986c8bcdde8048bd89bd07fa4dbb0c5c : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5ead8c193644a4bbbc91ef9825f5fed : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d04c5e2927324e9794ee2d1ecb950d38 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b3b374983e24bf99ed1eb4d5b8d0f9a : SUCCESS in 10m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8985ee45f30456fa738046b9c6bfa31 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78f500448f274387900e5a689eec3eb4 : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7751fd03360c49878f02cb9f32115ce6 : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/157b6212a4ba417bbc2c02d74bdcc538 : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7ec88816ae24259985e6250d48b7613 : SUCCESS in 7m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c30353a61eee4effbe08a1361a9655b4 : SUCCESS in 11m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4abc633f5734d07b6cc1cfd5f305502 : SUCCESS in 16m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8d4f2c179e4f4ed6b17f86aa9bfd45e1 : SUCCESS in 6m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/849e640d321441cb889722268880a46f : SUCCESS in 34m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3c3693de3644913938162daffa27930 : SUCCESS in 13m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee5c7ff0770f4f97be9b3619cd2b7864 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e45fb8020f6b4a1a86e0b652328f9128 : SUCCESS in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c4c40a3c3cd648f6a75fcc414a94b59f : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/936911b5f7ca489aa5c34a9b6da1d590 : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/664a588200c146e2a2b820658b14653f : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e8c73251044471a868dc2669e088f8e : SUCCESS in 59m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b1de473388a4e71bc0610218dce7b9a : FAILURE in 1h 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"754b5e948dc8cfb8a35451d406b5cc77d7648d56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 11:56:04.000000000","message":"Patch Set 1: Code-Review+2\n\nPart of the RC1 milestone","accounts_in_message":[],"_revision_number":1},{"id":"eb70462e2bbbd5d277f120c8453318f086b9c161","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-22 14:01:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a1740e1c03288f4565dd12571d4024db7d29e82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:36:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fe0cd016ba8a422b9791cfb655f2fdac : FAILURE in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"60aaab88b2573b0e88ae01e1b7c2e3156d44553d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:54:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d36d908e73492a960245b269005374 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c101cdc638e542929f3e329e778f8ce5 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa86e113d25349eb959c487e3c3f6433 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55b05ae439c4a4a873befa538ef25f6 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f77a39c7024f09a31f5ee1f2c1a66e : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4a488d973114aa8903e815d073d90bb : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b46167e305d947d49cde2145088f0a2f : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea63de00a4c3430d90040895f82bf856 : SUCCESS in 10m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc9999d82d13462ca13389d39145613c : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0665cf1c76f343ba9771aeb3769f411d : SUCCESS in 15m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ca967ed68064de2b0e687235590f936 : SUCCESS in 7m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d71342616221411596d392513f62549d : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8378981f423423d8de5e395a353427d : SUCCESS in 8m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b37799dae56144ebb19f0c08f066e00c : FAILURE in 26m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/91e62387c945427fa8f9f34d2ef73b29 : SUCCESS in 15m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/59f88e988aec4a76b957e3fc6147960c : SUCCESS in 6m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d53edeb2bf6497a9c7610e960311e2d : SUCCESS in 32m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a36696156193443d8248611aa416dad0 : SUCCESS in 13m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f749c23dc5484900b68e8b62eb13fd03 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e324d4a5dad42f1b812f2cfd878d3c8 : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/05a6cc4260324f22993a2f981d673173 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b54cb6e42aff476d90ad14e03df227e7 : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/318b431006da4d52b452a2b71721066b : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8074565be16e4cf8ae864b90bcf027e1 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/447a378e443948d483d62b7210f79204 : SUCCESS in 1h 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"68e9dccb2c1c32ffac702d85464f146695c34ec7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-23 20:17:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3aed9a777a5d4420780b114aa833417ce11a0062","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 20:40:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9110b935be564715ab94aeaa09820f96 : FAILURE in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"155f4de75eef76ad985f98532c827e24c2af12be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 22:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467b5218b3e5428e8bf8d293f86efc6f : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630fdc7e367e434f9ef0b5e093f535ec : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d217f140e574971891daaa29db0fb77 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac42375211fb4a49a3f0d8b907e1ab90 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d445e4a7bd134598a8a83051bd553ae9 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd30373f39a54a3b96020d7d7a8b412f : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba115a7d267141fa868b0f305d6fa60c : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9507658c20c4865a76f58c1ab3554f6 : SUCCESS in 9m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef9d32e4099e496a983da956b4ad4524 : SUCCESS in 11m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a574b0956ee043f88eb344b87dad7f43 : SUCCESS in 12m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec090695d30d4074bd8fab92ddfc1879 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7df101d0f745448b94c5b115d990bc11 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1294d62a9f14087a70e049dc50dc8c8 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3fee5d754dbf4e56835b0089ffba13f0 : SUCCESS in 14m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f6d5e4cb152843958945a740c4365cad : SUCCESS in 17m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0aa924991e0849f1a9a7f9dee0542533 : SUCCESS in 6m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e0a101eb0092419bbaa4892c775b161b : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b1a1e5c3580b475e89d5952129d42a64 : SUCCESS in 16m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44ebb0e866234828aaf3c6079be30834 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c356f3f81da4d678441b7e20b9f1026 : SUCCESS in 16m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c2472247449a4272bf3f43033988c481 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bbf3a59d8624d1382370bb836b9c08a : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c0e813ce1b246a99eb2cbb41767f6e2 : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/896f25eabee14a308cdce1b02f9e710b : SUCCESS in 1h 01m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfd6decbb0ca4a5196362673c80d098b : FAILURE in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"47e1799bcaf561bc6671cce5c976506de499eab2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-25 18:58:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65b2dc0411aea82372feb288d0b879f569a5a396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:20:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c9e07fbbf9da40feb54a6e8a0b5aaccc : FAILURE in 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bba6c88e2a78a9fb1d0e0987b246a9c0724581be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b8a39bcd5a45c6956357ebdd010664 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a33c2e0b97944451b30dd2480478c648 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd591d8d1e57433b9f679da5a7883328 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab314b0e52f848e9be9c031d7cf63927 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/432bbabdbca24ed19eadfa23bb149523 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e9948e5bcbd43d89a8de2ca0700c65e : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f07c60d86de94547a5df0be2923bf9e9 : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f4d3d750831403db3167596c123c129 : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3425beffd66f464485511398cf0687a3 : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/77abadfd34f448d0b7857ba04c822512 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/808a063cb65b4a78b6233c24164346ba : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/da507b4ba6bd44319242dcc9b82a38b7 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9d58a32d230485f873ae233227ad35a : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d8e0f53e30b149e89362b400182b8bbd : SUCCESS in 11m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/224611f42cb94798baeb82fa0f7a1823 : SUCCESS in 14m 19s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b4ab5d3b793e4e3d8132960bc7054876 : SUCCESS in 5m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5d7ed8cd96549eeb813ebcc9805b99c : SUCCESS in 23m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c40d68376495447390bf639ddd48c84e : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53a6e51b82f3467c9c97a8308f8a93f9 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcda1f06380542948ef1cd500c2397af : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d4093c746924d59b04c2d68a8f5f08f : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c91e1399aa14170a04cfbe53f2d2870 : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8bc7bbc09aa49b8a223db42669d8ab9 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16c589af0df442cba95f6c20f603c330 : SUCCESS in 54m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/883e9f2db3af463ba890dc1e1b958d0d : SUCCESS in 1h 23m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b9ff6e2374b5f84aa4c067aa37db51a5e16195e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-29 14:12:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f97bfc46824651e41f2b59e1b26c27b7804101b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 14:13:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d505ef9e050c6e00f2575b1bd8d11e3d2f87c51c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:06:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817016141f204dfeb90be6507f5fd9db : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c7b31346cc4dc1a3e786da4bf53a59 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9254c45c414d5c93f125343ce64c45 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/092d3f6a558d4091af4804cbe06fd49a : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3eb4d2b60574420381776f8d681df41f : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44fa53e04c9d46919bc30d75681c9906 : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/84836e72b4c7496c96ee20b4e15938b6 : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1327f8c0923c4407a0e7067ff343c83d : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c91002b01d8c47f7a675383e5c071f72 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/23b950d8cd65405a8fcae386219ce1af : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84f4dc5b3a034c18bea2a231832313da : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa336549d08d410391b438019af219cd : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a808df8d34d45c1b2514e6f19d617b8 : SUCCESS in 12m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e660c80f231d465681f7f16693ea7757 : SUCCESS in 12m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7d3c9b5050394adab31e859ea9358fdd : SUCCESS in 6m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42769d309ef84737899064a700ee42d3 : SUCCESS in 26m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f79f13ce91e4d80a4a89c134a601571 : SUCCESS in 13m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abecf062c37f4277b101be34e543a119 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2101a92273fe4afa821f99786d99ce21 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8c70797947c49b1b3bdab6d1236fa37 : SUCCESS in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9933a05afdd0401e8d7fdbff55259c14 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4b2cd71846234700b4358eafc0e4cb91 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aff142689b1e4ceba79d1ea8f7627b65 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4532ba83717b4a098ef29f902aae70bd : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"edf7b538ed1d4bda84205a22ed29dba11e70176e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:06:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2a052786f96e401b2d78e592601cb3f2203116d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:08:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c4f5ee8efe441cbacae367c55627dd : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdd39fe43e9c0fd615e7cbb01de39a514cffebb0","revisions":{"bdd39fe43e9c0fd615e7cbb01de39a514cffebb0":{"kind":"REWORK","_number":1,"created":"2021-03-22 09:49:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/782136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/782136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/782136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/782136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/782136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/782136/1"}}},"commit":{"parents":[{"commit":"8559da8d6b2a0f8d338e3ac6b12b3011fb95402d","subject":"update constraint for python-openstackclient to new release 5.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8559da8d6b2a0f8d338e3ac6b12b3011fb95402d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 09:49:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 09:49:17.000000000","tz":0},"subject":"update constraint for ceilometer to new release 16.0.0.0rc1","message":"update constraint for ceilometer to new release 16.0.0.0rc1\n\nmeta: version: 16.0.0.0rc1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release candidate\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I65f4bd098cbbb8311dbf78542711edc0888b5358\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I297916c3b4a4c6756350b21868f1d57c26937208\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdd39fe43e9c0fd615e7cbb01de39a514cffebb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdd39fe43e9c0fd615e7cbb01de39a514cffebb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
