)]}'
{"id":"openstack%2Frequirements~783076","triplet_id":"openstack%2Frequirements~master~I001e534ef7c64c96c14d6792a24d727eb5e77d91","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I001e534ef7c64c96c14d6792a24d727eb5e77d91","subject":"update constraint for heat-translator to new release 2.3.0","status":"MERGED","created":"2021-03-25 17:59:41.000000000","updated":"2021-03-29 19:11:54.000000000","submitted":"2021-03-29 19:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783076-new-release","meta_rev_id":"b293ff851f9aef969960c22eabdf41751102b1a2","_number":783076,"virtual_id_number":783076,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 19:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-29 14:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-26 08:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-29 14:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 19:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 08:35:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-29 14:13:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7679acfe90f9a48ecff2a0bac3e5bacd78f76238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-25 17:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd53f1a1d4de8a480ab4c9accd763f0d0ff00fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709efc55754f4f0fa528744bb8844ddf : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58464fcfc8e945c6b3d5dd2384f12e39 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40a74e3f810f43908dcd91419f7e6620 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4a461a9cc494e29a34e856f09647afb : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d62130f171234b7491ccc0c2b9a3e88a : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c608e34ce96440309e1acae89d55f050 : SUCCESS in 6m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/738c1c73debd4626804553b6fd64fb39 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b598d7d2cee747cdbddfb7ce3775478b : SUCCESS in 14m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c7196c9dbf446149b4f78c6fab969dd : SUCCESS in 13m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/474d92b972974735b09a0c9494749d11 : SUCCESS in 11m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7bbe9f634afd4bbaace9da5e84c2daf8 : SUCCESS in 7m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54b0a97fcd1f4bd89f24ee419433abe4 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acae820997254137bc2e60ccab9ac6f8 : SUCCESS in 11m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2a7fb7cc46114100ae67473a47554caa : SUCCESS in 11m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e19430bea2c941958b1edac9352d23a2 : SUCCESS in 21m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/6e7f9c455e184d2bbe10f0e292b663ca : SUCCESS in 7m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ab8780ee6a2478db35e3de5f63f3bb4 : SUCCESS in 28m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e957bf2c48b94ca58c67f2a587a5ce4f : SUCCESS in 14m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ce7e6d55cc6493b9dba25b7033a25f9 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d60f5bee7c54b2da4b3bfb083ffe562 : SUCCESS in 20m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7451b3fa5e934bb49f8f29420e51ceba : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08ee1b6e59f9489681e81f04967338a9 : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f72704516b0b464aad844dc39ea55ac8 : SUCCESS in 14m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5feafe993b7f47a68901f2839a78cd5c : SUCCESS in 53m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88255541cab44676afed4f9cbbc52e37 : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":1},{"id":"09b7d0ba2c7e1c3b9ff5f51e2f52149408026edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:10:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"488a584e561dbdaabfb5aebc6be1ce8b67415203","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 08:35:47.000000000","message":"Patch Set 1: Code-Review+2\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-March/020977.html\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-March/021163.html\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-March/021238.html","accounts_in_message":[],"_revision_number":1},{"id":"f47f0d6cb14677b3059b538b3c4153ae74405391","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-29 14:13:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c4ced1c444e6ec6e856137fb529192d6f036555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 14:13:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8f468f077b71ef7087f7c04febe99d1c046ff51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:10:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6002a97111c48ca8f9deb008cf47226 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf3ccaa6a374943a001200119e05d45 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f847a8294ab24dc086a2957334ed2408 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/879b9737613b45a7855d9c04872487c7 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49f9a05563524bedb67c7e9fdda907f1 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02b5c4581d9c41b5a90f5b69cefee222 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da84987309c04ed2891c3c8eeed64253 : SUCCESS in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/125711369fdc412cb48c83fd901d8984 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7d3fd8fcfed246b6b5523c07de5cf940 : SUCCESS in 10m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d51c701f5557420192a92bd408395f18 : SUCCESS in 8m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9986db36f664e51a0d1d6270f4cbc67 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb9bfe72e6094485900526f0e71fbe7c : SUCCESS in 10m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a9250602aeea4d43880e787eceab7baa : SUCCESS in 10m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f366aa2f9c447fbbbdccd52a192cbfd : SUCCESS in 16m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/be00b216a21241029cbda0b4d5ed0a65 : SUCCESS in 6m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/98a72f29faba45a5a32e5f4345b53671 : SUCCESS in 33m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b8480ada3b5c4c0d8c74b058cd2a039c : SUCCESS in 12m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/619e5646e8bd4f5ea9a374ebf726bd1d : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b35c3c5a95a4488bbddb75c338f0aa6a : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6fc19852a954bbcba8e5f2c3f928823 : SUCCESS in 9m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d68d25ee09447e680420deac43dc1b6 : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1dbc7dbda8648ff95798454100be0c4 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dca91871d844cfd8dfca1945009b67f : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d91ddd942b2943a0b84a804ea180b9dc : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":1},{"id":"38c28c95379080084035fbd10ece4dcc101d304d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:10:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b293ff851f9aef969960c22eabdf41751102b1a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:11:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60aebe3f72b4de8a8101249e65ddb7c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2419386b1dde38b33a546505cd03fc834d640ef6","revisions":{"2419386b1dde38b33a546505cd03fc834d640ef6":{"kind":"REWORK","_number":1,"created":"2021-03-25 17:59:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/783076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/783076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/783076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/783076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/783076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/783076/1"}}},"commit":{"parents":[{"commit":"8559da8d6b2a0f8d338e3ac6b12b3011fb95402d","subject":"update constraint for python-openstackclient to new release 5.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8559da8d6b2a0f8d338e3ac6b12b3011fb95402d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-25 17:59:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-25 17:59:39.000000000","tz":0},"subject":"update constraint for heat-translator to new release 2.3.0","message":"update constraint for heat-translator to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Yoshito Ito \u003cyoshito.itou.dr@hco.ntt.co.jp\u003e\nmeta: release:Commit: Yoshito Ito \u003cyoshito.itou.dr@hco.ntt.co.jp\u003e\nmeta: release:Change-Id: I9687062532d9f31be17d194f8922ab0d3f5d09b5\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I001e534ef7c64c96c14d6792a24d727eb5e77d91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2419386b1dde38b33a546505cd03fc834d640ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2419386b1dde38b33a546505cd03fc834d640ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
