)]}'
{"id":"openstack%2Frequirements~784355","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I6430c387d6bb33541f01a75518726055f7c51670","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I6430c387d6bb33541f01a75518726055f7c51670","subject":"update constraint for os-net-config to new release 13.2.0","status":"MERGED","created":"2021-04-01 09:43:49.000000000","updated":"2021-04-01 18:31:45.000000000","submitted":"2021-04-01 18:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784355-new-release","meta_rev_id":"58c28aa83724f2d901d089868754a528bee3dcd0","_number":784355,"virtual_id_number":784355,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 18:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-01 14:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-04-01 11:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-01 14:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 10:17:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 11:15:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-01 11:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 14:42:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e499ec82746b1c1994afd77b6ab341a22fd15b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-01 09:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70e6bca1ecfc02b410253e4058e43f963a326c46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:17:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4b2246ae41734e5fa967d630b5c96e2f : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9d640c0f261266725676a9ebae622f2a15dd0d49","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-01 11:15:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c5c73caf9915d0e8dd7258d83ea75acff3a5b72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985f8bf26785471dbc28ba59c9930c8b : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17bb2d711e2a42529dafb659eba54be8 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888a11de031e41cca55694b1ab0bd0e3 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa2258de7c5e4a7ca909cd2fff79dee4 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d112af221e2c4f58907beb07c541e556 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9057a731b3bc4e56bfb20c60d03e2dc0 : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c485d231d65467eb1927861861f3ad0 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/69485767055747edbfa411ee2ec2792e : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/615d61e0ffc54f4e8a0a600a25a81262 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2ef5db27758e417fb112e87ba1d8124a : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/443ccfa77a594748a4c956ab4fcc9d55 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bdb3cd1b24c54015877ba9eaa81d7963 : SUCCESS in 10m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9014e6af49ba4ee991a228de5007552d : SUCCESS in 13m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bf84418620b42bc8d92f8fe820a47b5 : SUCCESS in 28m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d3298e723d7d49e28ad354925b3de747 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ac31f5b09964e64b8ffb626089fa3e2 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f157c8e41c647e09b2ffa52a577cc39 : SUCCESS in 20m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/338d30011d4f492e999f2af157662017 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7a9b21d211b42538226f6c616e2f10c : SUCCESS in 7m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37b1160136304e62a8b0989a11769bdc : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12eb2ba3493045b4bfc66bd0895cd0d5 : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87aa35a3e1f345ce927f34bef3bd6f18 : SUCCESS in 1h 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"41185b0ce7c2303358b1be81f132d432689b7459","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-01 14:42:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e4fdc61147ba5f5d72ade35e6093e9c136eca1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd3623107c1ea6e7b6932eeb831567deea9e6b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:43:18.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aec662890bac3a0b7671afe0199a4ea709c9630b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c81d05457440c4bed47c963241f79e : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dbc4ee9a8a44eb1b5759e7160395647 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d54bcc63f1cf4462bec884cc2c74084b : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa2f09a878324127b97b9688becb96d7 : SUCCESS in 8m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d4dcf0f760d4ae390e68193e9cc41ea : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/076a4aeb02934235bbc1060538e782f4 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50cef0bbea754164b1feb3abe76c4c65 : SUCCESS in 10m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/80dd7b1c5c9a44c4b383cb1d0bf0b1c7 : SUCCESS in 10m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/88ddf89c5d1c40eb9a0286b84874302c : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/74d85567c44540728f81d537518e8605 : SUCCESS in 14m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e72cc70d9fa04da2bbe78fb527f892c9 : SUCCESS in 16m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/19f9cf769ce94110b627be3f3b8c61ae : SUCCESS in 10m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8aa1cc6298a5413a96fa50157625325d : SUCCESS in 20m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a873c754203f4b2c91ab580d59f294d3 : SUCCESS in 27m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a48e48c1b70416fb2df0665f1880a48 : SUCCESS in 24m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a5a646e38f846c49f6fa47cb5451063 : SUCCESS in 7m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/108fffee6eb74eaf82b06de488cf7e24 : SUCCESS in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a7a6fc57b0bb4ee1a4da2827d5d5bcf6 : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56740b7d3d5142829b646e4ca397ed39 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8af605cce6a4f919274ca9da57a72bc : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5734ea390067478085b681ef30dd2e53 : SUCCESS in 45m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/708f40d7777e4226a966dec994b66217 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"58c28aa83724f2d901d089868754a528bee3dcd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b494ecbb39e4be7404c9051fe639f4af185275bb","revisions":{"b494ecbb39e4be7404c9051fe639f4af185275bb":{"kind":"REWORK","_number":1,"created":"2021-04-01 09:43:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/784355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/784355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/784355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/784355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/784355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/784355/1"}}},"commit":{"parents":[{"commit":"628ba9dff2189b5392c8a9b5c232c1f261803a9e","subject":"update constraint for neutron to new release 17.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/628ba9dff2189b5392c8a9b5c232c1f261803a9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-01 09:43:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-01 09:43:46.000000000","tz":0},"subject":"update constraint for os-net-config to new release 13.2.0","message":"update constraint for os-net-config to new release 13.2.0\n\nmeta: version: 13.2.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ibfee5879fb7a568009ea8b1b6216d5c0f72b210e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6430c387d6bb33541f01a75518726055f7c51670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b494ecbb39e4be7404c9051fe639f4af185275bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b494ecbb39e4be7404c9051fe639f4af185275bb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
