)]}'
{"id":"openstack%2Frequirements~784357","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I7c7320507479facc65a4199ba6ca7c9cce28a0e8","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I7c7320507479facc65a4199ba6ca7c9cce28a0e8","subject":"update constraint for tripleo-common to new release 13.2.0","status":"MERGED","created":"2021-04-01 09:47:49.000000000","updated":"2021-04-01 18:31:50.000000000","submitted":"2021-04-01 18:31:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784357-new-release","meta_rev_id":"e96a61e217692a98dccf0eadd747ed1c387322d3","_number":784357,"virtual_id_number":784357,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 18:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-01 14:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-04-01 11:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-01 14:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 10:28:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 11:15:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-01 11:16:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 14:42:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0caefdd055ecb4975456343bd0acd43c37edd13d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-01 09:47:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3368666a8b4dea37face07d7715d1188e0263a28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:28:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/276cd817538c4705816178fd3dc9b7d9 : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3609d90fbf46cadb8757bbdac56ca6f69af7577f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-01 11:15:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd4b5ca69500ff3b3ad511521ff244d69ad1b240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:16:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a9d5ed672254650aef5dcee9f6947a3 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a30a3ebacf024d15bf02cb3d34191734 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b0d6c6886b74c10a660444b64f2906d : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34aab9efeebd4fa1b1f5e7f2db563106 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68339420d2e44357924c96f2175ef2b9 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7593ddc816b4cbda84586dac5a23dbb : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58e4643b742a4a11afdf1bd5d817a861 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c807314cda324715a77b4e50aa9d7a23 : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4989895baf114d588a27983d1f46037b : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1041c3d70162466fa2871dcf325272c0 : SUCCESS in 18m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e09354d8cfcc4d5a95d2819150fb223b : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/05d91d69d7a34a23a174c8344de3e30e : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51308cd2db5444669e82a6cbcd91881d : SUCCESS in 16m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08902dad4c5042b59e67fc31fffa04d6 : SUCCESS in 28m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c9bc111146c4d289e381b70e3a01402 : SUCCESS in 13m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2fda0b167923440f8882cdaf44d260a8 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/368cac47d6644d5db3bf0a0467a54101 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f30affb57e614aef84176c382c01cf09 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/520174695a324ea0befc25ab882684bb : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ecc56401d2394479acf78bedb3e21ed5 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b87185c054544e098c768b52792b622 : SUCCESS in 54m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/762fadf2a4794d0eaa81c4b35ab19380 : SUCCESS in 1h 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"dabc096f32970b6f262185f3f237d939bb97eca2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-01 14:42:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48697c7b89d56045cdcb51e84f891de42eec52fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec9de0dbcb7836582cc8095197114ceff17d26bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:43:35.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e32971e843ba2bb5446cdd1c9d30f66d50d394b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41eae51e05a444eb85a8c46769c046be : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc573279c42e424aab5ebd60a47fb125 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83f1fc0c68a140a9bc9c9858f34cfa15 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1783139de1964ba8a1d01bacc40e5760 : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36cb40a454df475db547c68717587911 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce093ef1a3ce4997b6af94f4b1699b4c : SUCCESS in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13b149cb34b74df6b4c0fd4de516daf2 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f1e9819b79104fb7922a97307bd7a48b : SUCCESS in 14m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4b687cbf97e4835b1cd79f9a4604768 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/12aaa7d7b1164e2cb78c77330497a361 : SUCCESS in 15m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03958dfd9d5c4c6eb664e0afe5688101 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bfc27e742da14d179fe0be508a48e5d3 : SUCCESS in 11m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9c630f239dc5462daf8232477255f31a : SUCCESS in 25m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15633bb4611442d099c9404cccf64d93 : SUCCESS in 32m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/476ecb4754b74510970df9ca35cfc449 : SUCCESS in 14m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c64abf455459439c803e28874207e370 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/858854a1c1304433be820f452cd7be97 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1b99493c03ca42a081f6de336da2adcd : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c8deda70b3a4684b5746d6403088d1d : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5e3d30674aa48ffbf75840f3fc77efd : SUCCESS in 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaec6e8bde354cc68f9578d87b41f30f : SUCCESS in 40m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/488112530c4749b49d52d398d6fe94b1 : SUCCESS in 2h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e96a61e217692a98dccf0eadd747ed1c387322d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79d0fef146d5c12dbfab1124ce5b5e50a5741e9f","revisions":{"79d0fef146d5c12dbfab1124ce5b5e50a5741e9f":{"kind":"REWORK","_number":1,"created":"2021-04-01 09:47:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/784357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/784357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/784357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/784357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/784357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/784357/1"}}},"commit":{"parents":[{"commit":"628ba9dff2189b5392c8a9b5c232c1f261803a9e","subject":"update constraint for neutron to new release 17.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/628ba9dff2189b5392c8a9b5c232c1f261803a9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-01 09:47:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-01 09:47:45.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 13.2.0","message":"update constraint for tripleo-common to new release 13.2.0\n\nmeta: version: 13.2.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ibfee5879fb7a568009ea8b1b6216d5c0f72b210e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7c7320507479facc65a4199ba6ca7c9cce28a0e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79d0fef146d5c12dbfab1124ce5b5e50a5741e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79d0fef146d5c12dbfab1124ce5b5e50a5741e9f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
