)]}'
{"id":"openstack%2Frequirements~784753","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iea5312c71a6bfcec9ee1ed9bf2502fa722b678df","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Iea5312c71a6bfcec9ee1ed9bf2502fa722b678df","subject":"update constraint for paunch to new release 5.5.0","status":"MERGED","created":"2021-04-05 09:15:48.000000000","updated":"2021-04-07 03:37:45.000000000","submitted":"2021-04-07 03:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784753-new-release","meta_rev_id":"9d16ca9c6c0ec0153d03e3ea566b5103f5b4bf46","_number":784753,"virtual_id_number":784753,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 03:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-05 14:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-05 14:27:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 10:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 14:27:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"73f0b0be86b0363a6fcbb4c72a54c8778d17a6c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-05 09:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5df2dd5251d339491c6ecd3cc3bacab0d3763dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 10:57:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097903b878544174a2d0cda8ba9353f9 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1935ba354d6e4b108b21e149fb2729a7 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d951549748d4f2488dd6360ddaf694e : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a29a032fce645c39e8e25a4dbe6c7c2 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a260ea3bb1247868bfe8c1f90b8e8a4 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76922f29e9a94d7595b4c8869a76b75d : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36d5d138638b45c3be31ecc07f1f1843 : SUCCESS in 8m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29be136821694eb491697778f5464c53 : SUCCESS in 11m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/39949bfcf51148c7ba09c1d39afe16e9 : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/61ef925e8a5043bc92f9940c914bc926 : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/458ac3d2a4184a39a03146c35feb1d90 : SUCCESS in 5m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/566417030cbb41eba8a9a8852695a8e1 : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bb7dce20f3d64028b886791e2a58db20 : SUCCESS in 11m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7ed1d760aa6f4bd488db234e2f129092 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ee9c9b939e1420c9a7fa558deb60258 : SUCCESS in 6m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5954535d15db474caba0fc8de8013b5e : SUCCESS in 15m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a4afd90ec4004ad0b58ef18ef0781fe0 : SUCCESS in 13m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1ee09e3e6e4a4030824a2714dfb57d53 : SUCCESS in 26m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2b7f6b9f02a34846a64e78d18a8a4e76 : SUCCESS in 24m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1c850c4f646e40ffae143df887ed37d7 : SUCCESS in 15m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49ba7b898e2940028a8e7a9a8208dc5d : SUCCESS in 19m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2f01cffc2a544268a327f350acbe6da0 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d982f2a3243844feaad2972ebbd1833a : SUCCESS in 18m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b0b43a5c9f749e7a8bfd639034ca656 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0fa8fcf642ef49048a267de3c9b74a78 : SUCCESS in 10m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a36648cf1e5345f5b8adb7e80ff0a112 : SUCCESS in 10m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bdbb2edeefd41a29c0d6bc0b421127b : SUCCESS in 1h 00m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb4de3657504494fbcaf37ce6d82e220 : SUCCESS in 1h 35m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/796e5e7d95a747df8f9e6d460cea0d2a : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e4d49866f8563a3a0952dbf6a5df4d45b219b620","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 14:27:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"722c39a216ec4c5a4ed1b79f098b4c633b0f0c1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 14:27:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a4f1c636cb6ff819decc39be237df9fbb4c899e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 15:10:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b8d8e7da1b44a78489901fd8204153 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1644d37a4786492aa05956c492136912 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647c066211454188be6284d41525f148 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/368ed1421bfc4faaa4f68aaa88d647b5 : FAILURE in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cf990aa172a4193b1fc90c3f2647bfe : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d49dabdea3d3491f85c20fd24a51f4de : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/87a87198a4464331beca8a41d5dc3f2b : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db9563e5a02b4118ae56eb0478fc9355 : SUCCESS in 8m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2c2b2dbe85774391a0eb8ac958577cbc : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1630daabd5a647a9a06977536474c4f6 : SUCCESS in 10m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e49abe2e5b5e4f7ebb6b76d981e55516 : SUCCESS in 5m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/077655806c684c24a8fb3bb71b120869 : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7c41412097c2425480088d178f8e2d72 : SUCCESS in 10m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/71cdc4d13e504d328c10c83800e39317 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c72dc8f89c8f49d6af47df1b35dd6cd8 : SUCCESS in 6m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0904f099cb8543a7be0fd7f8c533d165 : SUCCESS in 14m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f63271d463864fa69c3170cc6f5f2582 : SUCCESS in 19m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d9458568cb2241b3888b147b614f4cb4 : SUCCESS in 20m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b84557f080774702a5f3e304b1686836 : SUCCESS in 22m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/82637068cd8e4cdc847cfa4a5b65be29 : SUCCESS in 14m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/188c92e809d442e7ad75902ef0fe3fdb : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/306bdfa5b99f46e4b857c973273be487 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf2db6b7fbf942428ff755238d41a4d7 : SUCCESS in 14m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2985e07b5e045dea78b11da3c921e8c : SUCCESS in 4m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3aef7f9b6b624a7f98a8098e9a12cc0f : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f8ba8a5a9ca41e08eaea980898c780b : SUCCESS in 12m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0dea090989b4c21ad57c429a950e48a : FAILURE in 20m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6660c3497b1d4881abf4f4ed2e0658b7 : FAILURE in 13m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/41c40b4d053149798e42fb68843bb370 : FAILURE in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"509352fbcf71c9d5bfdc922e6a9aee8c1c7c7192","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 15:55:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0cdd5882f04d6e0f5010f3c564ef62da49f03d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 16:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c591fe2e59440939a01e89efc100098 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b80623774cc40348df375abb826dc28 : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cfb1d17a8244d1bae8244939b24122d : FAILURE in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4704f29af3534b04ba3890fbacc12a14 : FAILURE in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4435ebd60a9a4f56972532ab3299f5d5 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/133f1943aee1423fada5c82980f5efb7 : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/717f38afe86a474e9885de2e444f0f4b : SUCCESS in 7m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/349d06fe6eb54545abab640e0f05eccb : SUCCESS in 10m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0c4851ebf96d4cba80ab96ef8e9f2b20 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da63b303f11a44cc9bae875f17ae567f : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/405e29275fc24211a4fbf7878516d3de : SUCCESS in 5m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/827d2818a15d478dae6b99f73c1ffce5 : SUCCESS in 6m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/45aea4d800be4cc7ba6a1596b6f6830e : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48c122ebf964463cae6bd448ebe4f82d : SUCCESS in 7m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69517949db3c4c6886be72bb2555691f : SUCCESS in 7m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/68e0e6129b3347f19d70063424775246 : SUCCESS in 11m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6b17394a0c05466385d8d442dedb502e : SUCCESS in 14m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5c953a78005f421db7522fa3ff43a511 : SUCCESS in 28m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1306d86203504adfa0f0607a7ebf9a7d : SUCCESS in 24m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f787da05d5524e5680f44da832859a69 : SUCCESS in 12m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2fd18cc91880443da166a8dc96e2b766 : SUCCESS in 12m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a8cabd37df854799bd50ef55b9e3ba54 : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1651d3cf4bdd4dc69f61063be4959485 : FAILURE in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc789ed118114aa5b6ae56851d0dfafb : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/874acfca11c64094a6723b730f7e75bb : SUCCESS in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cdc0d794527d43ab8a84424944421210 : SUCCESS in 11m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/640fc00b7d374bf5a16ddf8783a66403 : FAILURE in 17m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/24a005a27cff4b5f807ef0d25403923c : FAILURE in 10m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ce11ec0f6ff42578b54efa1cbfe189d : FAILURE in 22m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1fcc95cbe5267cfdcc630ff1c1c0a48034864ecf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-07 00:43:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0bec66e66de3e41538d07f3b179b476470a57d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 02:07:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c189d63614408eb40b5119d8ea1f20 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43e997b5c04d4940922e615714d3c5ca : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f061820062cc4f8f813b350a714f0d0e : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4c071dddf574ee3a11b90fb98691515 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d6c3d9c996b42d8a5ec729e01a5b77a : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb29936d22404cccb52c4666b20aa296 : SUCCESS in 4m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1849358fb3df4878ab7537871bd4e1cb : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0cc5b563a7bc4e848039e28ba6cda94c : SUCCESS in 8m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6d587db40f4d4a37b35f54d064068075 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1cfe5e897852413587a8f352e912d8fd : SUCCESS in 7m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dcaf0315feec4bf2a0abda2cd091eaa7 : SUCCESS in 7m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b35fa38684f462e9baa9a66c3b09a3e : SUCCESS in 5m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2c091f657ade4938bde59da76dcb840c : SUCCESS in 10m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/28da5313863f4919b004e77f58bdca63 : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61f6d376230c4940950e690ecfa1f07b : SUCCESS in 9m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/96ea13bc5d5045ab8daa069c853a5012 : SUCCESS in 12m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bdb8bb7fd7a94037910d3aa04f5c4473 : SUCCESS in 14m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c484b9ff239a46c7be73a8c71bf1940f : SUCCESS in 33m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c93fac8f7134eaa8fd343e72fece6e6 : SUCCESS in 19m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d5e3c652b1c8496f8f02fc93e9150608 : SUCCESS in 27m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1fc79c7478d04b22861cbbf49acbe8d4 : SUCCESS in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d271a96e5e534d2c9374e7a93539e99a : SUCCESS in 26m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f6799e85cc340fa9d142a984877e91b : SUCCESS in 22m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d67c950f4bd49abbc291292d0f1e16b : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/49a3564c07cf4894ad400abab3787533 : SUCCESS in 9m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/da2c4eb5d514490a928570967a1c94fe : SUCCESS in 9m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44a70a91949d423da5225f01fa317ed1 : SUCCESS in 1h 00m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdb137da990c4c8b809890fe2c00405e : SUCCESS in 1h 21m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/55089f5d1f854f769e1c4c5947bbf39e : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f8f5ae72f410e924d360b28c079d9dde5d8a4cd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 02:08:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d08bf4c488638319aa5e9a1d0b689d30d36e4b5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:37:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8239bf6624243179ef90244bcdaa781 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b58060bec43454cb3ac7f7bf7aec0a5 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e175b42b29244cc989c5446d32a2480 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45a902ccd2954497bf7f774b34bea52b : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4065a4e42de345cfb5df487800e1229f : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f9560f0c30e4f40b466257b19792674 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba2ccab4594e4f698b96fb2558c33f37 : SUCCESS in 9m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d022c9f99d5c429e962dd2db5045392e : SUCCESS in 8m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/28d7d52f96a04c2f9098938a160e1293 : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5a7932f53a0e4c5a90b962c5ba84d160 : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b9541c4b1cc84697a1b83d1981da1a2a : SUCCESS in 5m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/189762f8fd2f42589746765ca3151bab : SUCCESS in 5m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/518ef5d5e87b420996a3a50dac38107b : SUCCESS in 9m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc1f325e23724e80985cafe3a253116a : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d86810311e774c08b6af9912a423b6d7 : SUCCESS in 8m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/80124ec0962c45deb8c19647c28fec89 : SUCCESS in 17m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/14d0e9175e8847168d0849bfe634b472 : SUCCESS in 26m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/02701c6f7aa34c22b026e2d2b11e974d : SUCCESS in 19m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/53ff46c09ec741a8b3971d10c84e3223 : SUCCESS in 21m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6363bfa3054440b295e2a4e6cfe5ffba : SUCCESS in 12m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/131f89f77fc74718a3e8873fcdf52c02 : SUCCESS in 19m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d942c232ff6e4ba4810e71e12eef74ba : SUCCESS in 13m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fb724907f694b7aac402afd45baa461 : SUCCESS in 22m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b66f75a0c7740e08752212e1c384879 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e6e91c4b0fb4d5389999f3eb452dd18 : SUCCESS in 11m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/08acc3c36a4a431c8c91de783cff8992 : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f858c20ee8c473396975d66c642ff5e : SUCCESS in 1h 04m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8115fc2b03114b388eaeb7c540c2c9c1 : SUCCESS in 1h 28m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/707ad0e0916f4522a299e520435a9969 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9d16ca9c6c0ec0153d03e3ea566b5103f5b4bf46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:37:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0c9bf1651f74fc34818dba3b9a459bca16fc7f9","revisions":{"c0c9bf1651f74fc34818dba3b9a459bca16fc7f9":{"kind":"REWORK","_number":1,"created":"2021-04-05 09:15:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/784753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/784753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/784753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/784753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/784753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/784753/1"}}},"commit":{"parents":[{"commit":"62208e2abf08046d182c6dc4d3ba094860222b22","subject":"Merge \"update constraint for neutron to new release 15.3.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62208e2abf08046d182c6dc4d3ba094860222b22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:15:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:15:45.000000000","tz":0},"subject":"update constraint for paunch to new release 5.5.0","message":"update constraint for paunch to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I7672f167a2279d7f00dd95b86ec83a8233e8bb63\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iea5312c71a6bfcec9ee1ed9bf2502fa722b678df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0c9bf1651f74fc34818dba3b9a459bca16fc7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0c9bf1651f74fc34818dba3b9a459bca16fc7f9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
