)]}'
{"id":"openstack%2Frequirements~784754","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ie11b07ed35ffb4f388f096c6ed73c8697b01ef01","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ie11b07ed35ffb4f388f096c6ed73c8697b01ef01","subject":"update constraint for os-collect-config to new release 10.6.0","status":"MERGED","created":"2021-04-05 09:15:48.000000000","updated":"2021-04-07 03:57:39.000000000","submitted":"2021-04-07 03:57:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784754-new-release","meta_rev_id":"e64eb2545bfd698696fb66ff64453e65885f66ef","_number":784754,"virtual_id_number":784754,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 03:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-05 14:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-05 14:27:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 10:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 14:27:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"36680db19e9367a385df043e4c746b25c898d3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-05 09:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c13a02b8bcf150afba0e6ac7453ea2b4a1a41e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 10:44:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3ce7d48c974b76be296ffa66c77174 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/955a06dde8864fbfb3fec3fba2b4cbcd : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cdd55c19d4f445b999007368a74d88c : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c603ad9b875f417fb25da4b712b97fdc : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/808e2ef1c22e4d54bba6bcacdce30470 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5360631a818483c880493477aaf8866 : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cbd7e43bc1464e5187466ddae55ec1f3 : SUCCESS in 12m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9256c71f790840eaa488212c4783c808 : SUCCESS in 9m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2dc34740500243c6a0eb23ba19b8c3b9 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a43abb9118f495097a2a308538404b5 : SUCCESS in 10m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ae8cc16bcd0d4810baf840e68ca301c9 : SUCCESS in 5m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dd385bd8540d4f52a42b547f1e5009b8 : SUCCESS in 5m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d1219cb1896d46f7b21979ffe4a6f682 : SUCCESS in 9m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d37dd4a59a384d118814917968ad625a : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97f97954d422481b9dbb4c601821b21f : SUCCESS in 7m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0208f2a1a59f489c98a97a2dec4a43ad : SUCCESS in 10m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/582ee63b80ce4bb69b0ae99bd8cdf01f : SUCCESS in 13m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/498f0ac4725b4c4abdecec17620a670e : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8206f008c61747ffa39da57421b0bacc : SUCCESS in 24m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a494e7a1ca7e45dbb88e98b555422aa8 : SUCCESS in 12m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6bf79de2f0064e5f8ed7da96eae55505 : SUCCESS in 11m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d53146b6c2b4035a008fcce74177e15 : SUCCESS in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d21039bd1a94655bc175970c40f551f : SUCCESS in 19m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31bec60949c242b99eedd206552f1672 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aadf9c76b82a4e8dba341e340b3bcd2f : SUCCESS in 15m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a465821ff75847f497285d57f35f02b5 : SUCCESS in 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bcc1c5b940f4356bfd6b96f677b60cd : SUCCESS in 1h 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/58a85a0ddd7d46bf8d21a3f9239a5c41 : SUCCESS in 1h 27m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b30b90f0a9ec430eacf276374442a3a3 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c73b5c9d2167ba800ae262b345d1dce93a307544","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 14:27:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19228421f8112754d186c3ef6bbe8c1d1d7ca91b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 14:27:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1749bf96ca3754d7fe4baa2af3a0b32a1e07ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 15:12:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8297a7ec896d4508a83dad7624e3627f : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/402269de903b45ac90e6fbf8e00aab00 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02d970fdd3ba4531b9c9d128bd248493 : FAILURE in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a32c83503c5d443282e48a9ee22216b8 : FAILURE in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/759977a1c65f4d208224436562a4eab7 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42fa414c81ca47cb9023fdf35da21099 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5b60e631b11c4d50a4aab446cf140b7d : SUCCESS in 8m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/365faae83e644d5da3fc6d50daaed8ee : SUCCESS in 10m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9fd14bcee26a4156a109deb0549db701 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f86604f691b141358d7d95354859e228 : SUCCESS in 10m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/af0d8c55639643949f683b48b1793c6b : SUCCESS in 5m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14422d419e0e4eafa76dd1ff08e9f30e : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/860e7ac2728f496cba047b68621ad4f7 : SUCCESS in 12m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4893f264a4743b08c771aa40c7338e7 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af4abfe5ef7b494791fddc34413b2b24 : SUCCESS in 6m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c3ca1addd9de41b69774204df694eda0 : SUCCESS in 29m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b941d7ca30241aabcb98daba3047479 : SUCCESS in 28m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d202ac1e59ee4a78bdfc7674f55bcabf : SUCCESS in 31m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cbf3aca386e343b6aa40054960a57667 : SUCCESS in 30m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f2cffa1f9c31452ea35c906366e17f18 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/508f7c1014474019987101c8694a9f90 : SUCCESS in 11m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5d23028f4c94892bd58ddca4eedb372 : SUCCESS in 18m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b43170a3b8e04e11a54420fd36208a91 : SUCCESS in 16m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d2e64d529534162b833a1de88594e73 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f2de0e488d6f4cf2af60ba6ea13242a6 : SUCCESS in 13m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/258c9b00121f435bb2481d156541d1f1 : SUCCESS in 11m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c2e3d7c36dc46b2954a1e64a77399e1 : FAILURE in 25m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4fda4666fc684a19b033a2d77b4e2231 : FAILURE in 9m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba55a5ad125c47d683c236004e9d5cb4 : FAILURE in 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2603e5c9c01f2301c62653636de3a96f7d3ace25","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 15:55:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b0862f7d920a0a25b917900eb5e6a2e4cbb734d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 17:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9e2b96feeb47db87233e6694c370ef : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebb1527726934e55a9b77dbda7cf750f : FAILURE in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/913f9280375f49abb7c0fd86c596baa8 : FAILURE in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/454cf71b8b874514907d49183845a402 : FAILURE in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfc5e4aa37ce42a387cb2e9a5662cc20 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4c678ee04614279a23ff99d5e0dc002 : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/95feb47a951e41d8ab27c3fab5495f54 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f20ffd64b92e477bb143b95e9f39a40c : SUCCESS in 9m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9ec1d7e8dc8142fcb3d42ae7c80ac50b : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c945340e6daa40d5992a5f3dbb39951c : FAILURE in 8m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/14de7d3bd1434a2ca72d397e4b0490a5 : SUCCESS in 5m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ba98eca1f3a46fca7e30082ee2a87d4 : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/63edeaf384b84e6380cf2cb63c14b9a6 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c42ea722c9b04621bf56df076040c9cb : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a12429623e2942e4ab3144be811cf85f : SUCCESS in 6m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8948488e0eaa4909bfac26ebc20670b4 : SUCCESS in 13m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8dea277cf6e491f852c6dee5f2c3292 : SUCCESS in 15m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/36260de1c3c64b588e0f50976c0b8142 : SUCCESS in 25m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50128486fd8a46a38aa0a5b3d5d862b8 : SUCCESS in 22m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f13c6b91e49d45329037465adde7cadd : SUCCESS in 13m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1bd7bc9e550f4ad99d03511e61073f1f : SUCCESS in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f27e40578eec495394dee0b337278f6e : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c42ccb00bbc74665930f6a2ffa1a1549 : SUCCESS in 18m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7bdeb056e3c41c8ac643f2667b95958 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f8547e3592cd442e8217a5a9b6d1190e : SUCCESS in 9m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/487c6b78aa034fa198322c95a044c869 : SUCCESS in 11m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f0a04a41a0041879264c1c1e080aaa9 : FAILURE in 13m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/21218da4d853475cb03b430ac7ad8f67 : SUCCESS in 1h 26m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a16c4ad809ac4969b83655545a9c1547 : FAILURE in 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"13937d6e274c8924c7fac692eea46b8fd514ec4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-07 00:43:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b703b23906f23205cb055b68e1c5361f8464b1d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 02:13:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664c1485359e4eeba4c5df461a4a2ad2 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6062ffc677004cf9b77552f6289a8828 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f315c170bcbe40688af2e6b183062c92 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e521856007346fbbaa7ec5a0adc488a : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/966bca18f2914979a17679b86939f8e3 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b272e62dbb4c476485cdeb26c0f2b011 : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/431fcd95b01a4ab698b30463a55a4766 : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0050d08424ad47d5a2a4b5a48a31d72a : SUCCESS in 11m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/958bd0fd4f914f79af39f0b2ea4b2824 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d89e96e5644b419e8fb2016ac3c96fd9 : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0d03ba8ec0274acda11bdceff33a272a : SUCCESS in 6m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/517466e52bd74df7ab3ad0840778e1b8 : SUCCESS in 5m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/10929bfdfa32499f8da1d8926e6457cd : SUCCESS in 9m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53558f2f0f1b45bb9a5e9d28eb627229 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14a71f95995f4e9b9fc8c8973025b62b : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dd9ccc9494af4e94847cd986af969f94 : SUCCESS in 12m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/551cddf3fe2f4ab08c05f7e6f50122a4 : SUCCESS in 14m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d0a95bb2fce9463cadb82e94bd667293 : SUCCESS in 20m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c95102f61f1c41aab9e974913f6bee81 : SUCCESS in 22m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a5acf7987b1c4837aac34b4444be4125 : SUCCESS in 14m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfd33996089a4679aed001924f40205f : SUCCESS in 11m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0badd13d595f42af8a5994db6cbd28e9 : SUCCESS in 14m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2fcf80ac82348d9865015b6d65e2988 : SUCCESS in 23m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0da735cc5be44e49b3f19d63f15f5ac1 : SUCCESS in 4m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef1e8471db034a67b394f13e1c4ee8d4 : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f36d4a082fd4091bce8e3208d390ca2 : SUCCESS in 11m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdbde40e25df41e69ff6a627da1c820a : SUCCESS in 45m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b29733aa9fc94a6e8e23ed2b84ea4df6 : SUCCESS in 1h 25m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4e7627b6c6a41b98d8e445f141c14bc : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8b3ad51e168be2abb74d755c9089e80afe933fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 02:13:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13166a5e6dca6de8107b306cd13ff181dd3de77a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:57:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ec1215df2243519b6408d6b46b777b : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08803b8d2ae04739b1a017458f49b6af : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b988dd7aab74a64901480ee873b9547 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e3be3f45ea44811948037bf03e44bc5 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6c38d6c18ef406db614338f3cfe4e38 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b32ff894ff604d0c93630cc168b6d0f3 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7822685218fa4639bb256ab78f37ce9c : SUCCESS in 8m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/280e47b7cab84dd5b6c1a4c509dbd05d : SUCCESS in 10m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/843d3556b3d64eefbcdf5ee6394b5913 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/37d9d1281aab45e9b9c6f0998f0f5796 : SUCCESS in 10m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/94fffe39c7df402f97440076f882caec : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4fd5ddcd56e6420991e146359db206b0 : SUCCESS in 5m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fd9082a5c6264497a5d4411f76b5bfbd : SUCCESS in 11m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8ea0842fe8b4db4848079a772a1bdb5 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b657e57df52421688785ac1e0180daf : SUCCESS in 7m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4fa288d144e44154b749a5834242d6a7 : SUCCESS in 14m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c1c1b333fb9441d181508809b959abc7 : SUCCESS in 13m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cd1a107bfe654dc99db9f5e72529f182 : SUCCESS in 24m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/717e7c6e412846ec8f16c81d223c68d0 : SUCCESS in 22m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b8db4474b3164797ac6feab6411e5c5d : SUCCESS in 14m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb6861cd9baf49fa86702d414f8deeaf : SUCCESS in 13m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2603f4db9d424b2faa97ecdecda764b8 : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ec6a0cac5fb4f16a2daddbc34ec253d : SUCCESS in 18m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b76ade100344e0d99f2eb339233a41b : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf6c3fcc68d84064a748fad84678f1ce : SUCCESS in 16m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0cc2ac408a404d5d90f04a8b3789c0eb : SUCCESS in 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/827abbf306d04f7ca77a56ad9e23d55d : SUCCESS in 1h 09m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1ff37aec14bc4844848cfbbc13f09e86 : SUCCESS in 1h 17m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b7f86e6e7614eb9b3005effb212ea0d : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e64eb2545bfd698696fb66ff64453e65885f66ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:57:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc05226f39ad6474a63077a58b7e75fd8a5d8150","revisions":{"dc05226f39ad6474a63077a58b7e75fd8a5d8150":{"kind":"REWORK","_number":1,"created":"2021-04-05 09:15:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/784754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/784754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/784754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/784754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/784754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/784754/1"}}},"commit":{"parents":[{"commit":"62208e2abf08046d182c6dc4d3ba094860222b22","subject":"Merge \"update constraint for neutron to new release 15.3.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62208e2abf08046d182c6dc4d3ba094860222b22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:15:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:15:45.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 10.6.0","message":"update constraint for os-collect-config to new release 10.6.0\n\nmeta: version: 10.6.0\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I7672f167a2279d7f00dd95b86ec83a8233e8bb63\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie11b07ed35ffb4f388f096c6ed73c8697b01ef01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc05226f39ad6474a63077a58b7e75fd8a5d8150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc05226f39ad6474a63077a58b7e75fd8a5d8150"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
