)]}'
{"id":"openstack%2Frequirements~784758","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I2e6ab215dac3ef93edfac41f38961b96ac552e01","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I2e6ab215dac3ef93edfac41f38961b96ac552e01","subject":"update constraint for tripleo-common to new release 11.6.0","status":"MERGED","created":"2021-04-05 09:25:06.000000000","updated":"2021-04-07 03:11:04.000000000","submitted":"2021-04-07 03:11:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784758-new-release","meta_rev_id":"ea07f34da9128924a6ce103564e9025a2af2aa1d","_number":784758,"virtual_id_number":784758,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 03:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-05 14:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-05 14:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 10:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 14:27:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8d612b89e4587985c7656525f666a8766aa1d2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-05 09:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9337160fc3064488f0aeaa1460e2b8f5cfae23f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 10:49:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a6f91e7e2d420e9be102c20182652c : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8b56fdb54a145e18fa6e8cda4150b4c : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e3258a967545229d5f186f99adbc08 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34728a1db50a4627ada3f83114f9e9c6 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f56d26ba4b764e6d946ff5750c275e0f : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/578db6b1dbe241b69ad99e5c345883fd : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fdef416df3c94fd9b5512cbbc79c91a9 : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/121a9841e8734b87a35edcc3c0583552 : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3d86da8b556d4d2e8057ae5ac42f398f : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b84f3adc2ebc43deace33a3a72d1bae1 : SUCCESS in 12m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/343c687e0846476fb2573710a7f8cc22 : SUCCESS in 5m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/73c985a980b94ea7a5851630385073ca : SUCCESS in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/22b3c12ff0d04ee88c3cb6915d520277 : SUCCESS in 9m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1ecb5eacceb4ea69e9f169fa5a63d20 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d2499934ab541bfbaf131d3759c8e18 : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/802f75b2ea7542b8874fb5017a8048ad : SUCCESS in 12m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ab01c2659375406b9d4a70d039c6a723 : SUCCESS in 13m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/39d938843f8840038ea818dfa4788622 : SUCCESS in 19m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f0874cfc044647b5abd301ec28ffe794 : SUCCESS in 19m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8df8512ddd14f5c87d22b8183b672ea : SUCCESS in 13m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d814f2dcb68e4c62821284186edf0d9b : SUCCESS in 12m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/00d4956573474375ab0fd03e319c0750 : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aed992f8afab417682613f8810bcf959 : SUCCESS in 19m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/782a5507f809495f83688def02647280 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e83747ffb2894d34b30fe4e22e12d8d4 : SUCCESS in 9m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe90e2b2b0f54a9fa2c22d1e6b085750 : SUCCESS in 10m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bec0f572336c4dff94ea135e0e257b22 : SUCCESS in 1h 00m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8392a81ccd34925ad7facf741348527 : SUCCESS in 1h 23m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8ca52abe36444613bcbec05755b9fe38 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dbd215962a517cd65ff65c556807e688559a9a6c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 14:27:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1379b210ff21a81d25f593d7244e0c0380931864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 14:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a40c26bf70c2f8fe3d310bcb2dfe494de4306cf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 15:30:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709ab03613ac43ac9a086326743ecc04 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/366ae73b9aa34f75882934e1084bf32a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7650d4877b704280b4f7ed05dcfe8d1d : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/878d18169ef343eba95346e8fd6448bd : FAILURE in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33214618f5e8417f9fcbc6fcdaa12caf : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/776d1bba3a2948448c76dbc72ba07365 : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8929723400da49958e78ca448dee698f : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59a89e5565154ef0b87cdcebe2a022d6 : SUCCESS in 6m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d5ff121a364a4fb4b2368a33c35b615f : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1d537deba5414051adaa662b8508adcc : SUCCESS in 8m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3d98f42a8ca4425a8156fa2369861713 : SUCCESS in 4m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ca036e8e259442c9edb7a9bcd306440 : SUCCESS in 5m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b888fac1d4e34393a8e4112347d682f3 : SUCCESS in 10m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5037d312480426daccf85c01092b124 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54485c1d55234c6cb8600fa029498dd1 : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e74d620aee5b4b0baba9fe48a0a5e132 : SUCCESS in 14m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b285c00989cd4a4486e2b949bbe42c7a : SUCCESS in 14m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e36fac12da554d499607f9d324a419aa : SUCCESS in 23m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed303e15add542f4bc6091b34bab9ed7 : SUCCESS in 19m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2e918c2dde7d430ba329094d97e8f6d5 : SUCCESS in 12m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32d731ef84634171b0c0e770dd49bd9e : SUCCESS in 12m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/58e07816b8ae412da4203567b7ab0e10 : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea1e4ed5c5e44f5292278a4f76e33c52 : SUCCESS in 18m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a44136627d6427e8ce7f1acc0ed4978 : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b4ba637479d54748a1c5721c217b4fc2 : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ad05c72eb5b94505a328947972ed9455 : SUCCESS in 10m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87f981f52e3d444ba6ec970c90abb5f2 : FAILURE in 15m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c595fb283daf4c9ca66480a66f470dd8 : FAILURE in 8m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e281f4e5bf141aa8c659af79b634c1c : FAILURE in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"52515b30531dbe245ef109e461ce61d39c771086","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-05 15:56:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29996c8f41d8ff22d15bfec75b8cd7381fb9f7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 16:20:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8773febc7ef24e0a991eaaec98c8d173 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/693ca588934d410b88e55967a1b99fd4 : FAILURE in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29bb65a0a4ee47d6ab61354928aea86b : FAILURE in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa83e21089604177b86bdfd8d688dd99 : FAILURE in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e4195635e194733a92178d0f504365f : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a7d32789c134584bd235bd9a1b08507 : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/957cc4922b4b4286a51e7429a7179b08 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e9fe3157e61470682035398dd4ab241 : SUCCESS in 9m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c37b1f2941214f2a83ec72419a85fa41 : SUCCESS in 12m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d39fdf9847204879887a2a179d4230a2 : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e46b261053854d36a2b26a80f942737b : SUCCESS in 5m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a13759fe8874b77a4c8793b4a7b5f33 : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b8b41585a0ca495a8f32414b5a64f769 : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1daff8c6edd9437cacdaf8bc39d79e3f : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfa195b4517447fe96dd9d0a3f5ca64f : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f486b5bfc6d34bf3ba50c09d613895f8 : SUCCESS in 16m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cc5965b642194e9c84c99cb029ffff3a : SUCCESS in 12m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c8a2d3c2827c4d3c83e38d071e71b82e : SUCCESS in 21m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8ea793dd9304e3686ada88f554517c5 : SUCCESS in 14m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/50ac088353804f6b94484faaca1f7f28 : SUCCESS in 12m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e69e566afcc4ff5b132a2636d380be6 : SUCCESS in 12m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c32c7f3dfce544b5a7a00aca6b85315f : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/560b3552ded841549f5168b160711f40 : SUCCESS in 15m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc88175fcfbb4d59b0edad75d7d70e93 : SUCCESS in 6m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3bb4306b65744c389fca34066099d01a : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cedd99afd74f45deab9731613a725fcb : SUCCESS in 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c2a7dd5bf647dcb9043c4ff088aad1 : FAILURE in 16m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e6854a79294e4542b262bc5e4a11b6df : FAILURE in 12m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f575f43f96b34d26bf374e8ee2812ade : FAILURE in 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"240c277a88878cbe160432e0ba311ca6ce527ef5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-07 00:43:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23b30195a490356f37695f9156f04ac5b82164e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 01:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e77f7d99e7a4fc4a24159b91745cf78 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e36e99367bc4b829d6a9c060e7a4fed : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17ce51bd47074b98ac14e4cb1379703e : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84e2284b71b94aa395f5759c99f50162 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d073b1fbed4e4c128172237756c4e907 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df5b4affc2354716bda61de6bf688bcb : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d92695f006474252a9fa7d6ddc6b676d : SUCCESS in 8m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28773d5daca64f2793f0b1d29b97de51 : SUCCESS in 9m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c82623755b3443deab9041e53fa0248e : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d87ce0baa4e74618b2c8cbfe6f503e6b : SUCCESS in 9m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f34e5ee7128143958a25752c41c6797a : SUCCESS in 5m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b344aa0a684e4e50b36aa9d515120c8e : SUCCESS in 5m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f52b35d1cdb7479fb78beb2b74e229c6 : SUCCESS in 9m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1928ba62d08f45558f63bc21a424502e : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48c292769d9d457eabd4e55012e6d2a4 : SUCCESS in 6m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b2f1bf75ee404380b58641e5c4aff333 : SUCCESS in 13m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ae8bff48f0a41c6909762aabbbadbc1 : SUCCESS in 13m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1c6ff78a9d4645b6b70eef8c883d7247 : SUCCESS in 23m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dc8b613a5c7b4136b64f70b870c0482e : SUCCESS in 16m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cbf4edd46a2a4c80b49b0336684d2be6 : SUCCESS in 13m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/610f8fc915134e62baa323293cea11b4 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/60fd669c438144b28476a529fcab5264 : SUCCESS in 14m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fb015f794f140c28dc7f795e75f14ea : SUCCESS in 16m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/998ade6d8c7144aa84b0f272398d1b15 : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/634e6f0bb9f74b77a250fe562300b4c5 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f5c37c93b4046dca40c701581dde56b : SUCCESS in 10m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88611b7fc5f7492e99755d81f075af1a : SUCCESS in 56m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9363bcbf371749708bdfab0bf7d0376e : SUCCESS in 1h 07m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ff966e87ea14cf091ca11ba902737ce : SUCCESS in 49m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4c46ce50d95a33d53f176b2e20b4f6170d14c276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 01:54:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cc9b40708d889dd6b132f19a5fcd55f873bac9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:11:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80adedaba56148c8bf23ecaabeb313aa : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c78622a0c584c7bbf4dabf8ca080c10 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0615a3320e24aec8f524e086c5ab600 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1208b7ed1c7747e88c45fa470a46efa8 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e21686f371564a8d972dab069f526ff9 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1a90ec6723642fa9467e0f49117205a : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/66b1d034a8164144a42b82fd30f73409 : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/45ad071d6a184400bfae19105992df7b : SUCCESS in 12m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c2f95cf7d98b4a9893aadf8fa7e2236a : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/250ae95582ab410f874ad7c042046123 : SUCCESS in 11m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54d04a9381e14fc283310664aae05908 : SUCCESS in 8m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/89859fed05624cd9839b208c9e0023de : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e455d94fd32e446f9acbbf2d9c859d10 : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f029c3d9d322472390171428433714bf : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cb280bcc50f4e6fba248793200ea249 : SUCCESS in 7m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b8d902b2e03440f3a571b327b94ff949 : SUCCESS in 18m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cb23df13fa64ba2b9bbc1da779a6f43 : SUCCESS in 13m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1afa14aa99284de7b7e5aea54b6d77d2 : SUCCESS in 20m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/334a4ba2a979458c8a3e4538c5946e29 : SUCCESS in 23m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/68bfbd086aef40bb9c65e7a389c4f067 : SUCCESS in 12m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cee7a7f2dfb847aa99eab4a6e806d194 : SUCCESS in 12m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec9e36750a334bc7810086ab5e8c5c50 : SUCCESS in 17m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/161099126b2e4c6cbe8469e494a92588 : SUCCESS in 17m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4f2c2163bcd479db112d33807ce7052 : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b4db92f67a7d47dc990058c39a76c944 : SUCCESS in 12m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a375902dc18e436bb6a4a84e2248e435 : SUCCESS in 13m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b89b8ceef054fb59a78ad1c56712160 : SUCCESS in 52m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e2a5c0ad15ef4256979d894f412d1f3e : SUCCESS in 1h 14m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/383f916d640443af98ce1fdd16858cbd : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ea07f34da9128924a6ce103564e9025a2af2aa1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 03:11:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfa0edbdd2fdde2a899479741ef618f131f09494","revisions":{"bfa0edbdd2fdde2a899479741ef618f131f09494":{"kind":"REWORK","_number":1,"created":"2021-04-05 09:25:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/784758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/784758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/784758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/784758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/784758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/784758/1"}}},"commit":{"parents":[{"commit":"62208e2abf08046d182c6dc4d3ba094860222b22","subject":"Merge \"update constraint for neutron to new release 15.3.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62208e2abf08046d182c6dc4d3ba094860222b22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:25:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-05 09:25:04.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 11.6.0","message":"update constraint for tripleo-common to new release 11.6.0\n\nmeta: version: 11.6.0\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I7672f167a2279d7f00dd95b86ec83a8233e8bb63\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I2e6ab215dac3ef93edfac41f38961b96ac552e01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfa0edbdd2fdde2a899479741ef618f131f09494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfa0edbdd2fdde2a899479741ef618f131f09494"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
