)]}'
{"id":"openstack%2Frequirements~785745","triplet_id":"openstack%2Frequirements~master~Id36985024db81eb9c9d6c5f3d8c7b10243029961","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Id36985024db81eb9c9d6c5f3d8c7b10243029961","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-10 06:22:03.000000000","updated":"2021-04-13 06:37:15.000000000","submitted":"2021-04-13 06:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785745-openstack/requirements/constraints/noclob","meta_rev_id":"814e5e0bae863af991a554cf45cad0400455dacc","_number":785745,"virtual_id_number":785745,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 06:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-13 01:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-13 01:45:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-10 06:44:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-10 08:55:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 21:29:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a34f4c1c159e90d69e702bf4ac10ec753dfd3346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-10 06:22:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e23306599812bb0ef301dd1122718fd0e0e85163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 06:44:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45a4c2ece3574dbb8a8ed6378e178963 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"58c0c8a1daf8c7a63fd5b5a2dac4b0b1da8028c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 08:55:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e14a05934c1424faf345b60a68c18d8 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64d312ea41c4ed1899d55976f4c4c51 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fefb55e58d984f1c8cdc2da0607e2627 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49a227c9898b4f809081ee470dd8282d : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4b9a07095de413dbea8bbc92b34faa5 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f62249957b384190b28b1e0bcfa6b879 : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5168d7194f5e4f5781d8d4f72bfda5c0 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6b82fa08a0e4bbba195846987b72f37 : FAILURE in 5m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2de4add3b1f64844a2127ccd62d87fb6 : FAILURE in 5m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c1e84a2d5753442982b54f6d7e62384b : FAILURE in 6m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da19b4dcbd7a4e42a812340af18d1e43 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/531822028ba648b8a2da5519c9fa1967 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc2c4aa41e90415cb924e0c5115457a8 : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/db4ee258d6d6450ab3e14c5a9b717630 : FAILURE in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fb8ea14b87954f80ac4a11e7a269bcfc : FAILURE in 10m 53s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/391acc1179ca4680a09748f6600f4dda : FAILURE in 7m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c43f0703a7d4e539f6f8b0bbc1fa276 : FAILURE in 30m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88c9bf046b4d479895d5d8bb62e6e9e2 : FAILURE in 11m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04960ec9e3864d4da4f7d0637f23d6e1 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b828e256933b4e2ba552f59eeb39aae4 : FAILURE in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a774e2df9b53429db7c99190a6f19fd8 : FAILURE in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6ddf63aa24042ac8c8530ccc727b376 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8eac79ed17364e669e401b316cf0276e : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eea96cf6faa445aa16c9ecffc34f166 : FAILURE in 2h 13m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9477295263f747faa25c14ca76c38162 : TIMED_OUT in 2h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5a90bd467c1992e539f75d22e83fd49b533ab41a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-11 06:39:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"448a5455953cee80eb5b437fca972a8f83d91222","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 06:56:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e4b1264d127946368cf2d2b1c4d314dc : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c75c012ec6cbeddf1f21668991d4aab8cd206aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 08:50:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0798f413b3244a2d9dd44be080521749 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8fe7b0963c435482d53146a6681314 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cbbf8e9abd048e2b27d21144988329e : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2927c920f6e448f18440a8125ae82fb0 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7556ec6b0794d9c8b642a8000127381 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/970c1aa9835c4b089011da28faa643b9 : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/604c4857d9ef49b1b627261d35941c9a : SUCCESS in 12m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1de38d0fdd4b41b78f8af49e9fd91d27 : FAILURE in 4m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99723843579a4732b8de1d4e23d56a30 : FAILURE in 4m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/74b0ba90805241869963ccddd702e7f9 : FAILURE in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5bb69c4b37ee49c78e22cd4acb8ec3d4 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/32f34083dbe540b1959cd9fbbd6e41e0 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9813c2ddd73a43c3a4b6db79fa85568c : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50626cb6dfc84318abbd37d62a4080c5 : FAILURE in 10m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/54ce486c190c4a23a4c881caa733cd27 : FAILURE in 10m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/327241e161944299a91f1830b3575c12 : FAILURE in 4m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ef46b4d4b8649c59fc73867108a34d9 : FAILURE in 36m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7fd313e3abb2408d9f05c61246f173f8 : FAILURE in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a4ae87395ad4462b04d9feecdbd6827 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/877aab55dd0741a2a0fdd70fcb94cf88 : FAILURE in 18m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/034c90e2b9254c67b8346a428517e792 : FAILURE in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd18b920810d440a87db059a14f23b5b : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a37fbf8bf8924e4bb05429d92d0cd54c : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/171d3e68fe5a4246a68c97fdf0d94876 : FAILURE in 2h 07m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce59b8a2086d4795a016a64300f74b96 : TIMED_OUT in 2h 03m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6cc28b72d513bb1226cd3551abd37566903f0b53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-12 06:26:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c74cde3dc6809fba848ce0a3cb7b89fd530d5a1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 06:51:43.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/242760f5a3a14232babe76812d1e5fd3 : FAILURE in 17m 09s","accounts_in_message":[],"_revision_number":3},{"id":"909b6a9f04d77c08e2402c404baf45555ddc22cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 08:52:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb32bd0db5ba471387d86e7330b3a5bc : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fd59d40d1a408cab726e9a8dbca0f2 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edd26c75a2fb4106b29753436e373f3a : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17de1780f554e86b37aacee387945f0 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39205efa2b1a4540ad836161c6624c27 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/567872fecf814da5a0fc1f5925c496ac : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d6dca1fa8f74143a9ed9420bca1e449 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58ffab21eeca4c57a3086ed7d3092ad1 : FAILURE in 9m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b66e93a721c94c49bfb4a9fac878007d : FAILURE in 5m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ab9dad49549040fa9e833e644084760d : FAILURE in 8m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf321fdca53e4c4bb74580b18aa4bd96 : SUCCESS in 7m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b3672a269984a35b14a80ea8df1a8c4 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e47429a4a6f0441d8506bba7b593fe20 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d6c4a19fc103466e8e1468181cdb3640 : FAILURE in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9a76bdab3ca4b93b19c8d0b27574a5d : FAILURE in 12m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f525318dcd444b38a8575b945c02c73f : FAILURE in 5m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61f2df683d3f497f8fe2eed1ff22e0fd : FAILURE in 30m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de2bd3bcf3364dc8a1ddda27607e07f1 : FAILURE in 15m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4a2c955dc73f49e7a357c62587a1c2dd : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b84e5886df714a5fb27b5fa2fef197cc : FAILURE in 19m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77d1b478ff7c4be18aee157e005f5b4f : FAILURE in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d0fd479a7494af3bd42037a83e9a5d3 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fcddd4ea8a044b48901d4f44079f8f86 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea238979fe804e9e9b70f76b7f93685c : FAILURE in 2h 10m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33932a8954fe4a1182c36801549d5605 : TIMED_OUT in 2h 03m 24s","accounts_in_message":[],"_revision_number":3},{"id":"10d3694179e5a091ccd61b5de91141bfa3cf4abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-12 21:29:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0125b708781f7cf28353bb0de785d265841e8d86","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-12 21:29:58.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"ecb040f163b36722da85aba074a5d73113bf1bec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 21:52:22.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f2f6741d582e43e99ccbb6a9138a215c : RETRY_LIMIT in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"59310869f868cc49d15d7f6adf27973b3c0987a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 23:42:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe51b054eb0941b5b29e664506ae68d4 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4060c984ae2b40a7bc11f9ff8531663b : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa764b8517424807b8b50bb8a9c59d7c : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db976edf7f3b42e39eb5d1aa4d77989b : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c102113fcfd1450f8c5528bce3eccb28 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f187dada126a49bd85e7a8db188b54ca : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b13f2cb5451c44ed8a591a2e9d8b5f0f : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10e294b065e544cb9023bf6803fa596a : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/303279aae1364db0a9dcad03c4aea2ae : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e1b670df39b94c5aba1241d0a8c6c5b0 : FAILURE in 6m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a126e09a55c4bd4a256e05997460efd : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9db13b15d05f4c17a6fb0ceaaedc7b6d : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebf2b3b39696475da368e5b41fdb3c3c : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c796e73962294e5dbc1f9d8b1303645c : FAILURE in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/49ef847cee264047a7b0906a53fa906e : FAILURE in 14m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f84752dd76bf4e689b3c456cc9653a8a : FAILURE in 5m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df63e69bb4c4423e998fd138645c248d : FAILURE in 32m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35fbcf4e1fd64351957cae4f956f8ed7 : FAILURE in 12m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34c62a5c5ffb48e896f5740a6dbb1c5a : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5c40723511643d8a4d380cdc6d154f4 : FAILURE in 18m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/287ba052d0584484ae3a4c01d43b6f74 : FAILURE in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff4e5ba4a31e4a168b18168b601ec6a0 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cdbb254b9d7642d2beba3731cbb0b9df : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/817f50dfac224791bdede0924bcea808 : FAILURE in 2h 11m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b827ac0a19bd4de89dd4471806358074 : TIMED_OUT in 2h 04m 15s","accounts_in_message":[],"_revision_number":4},{"id":"03e0e067079cf9510e5e4245455a7fa86aeb3025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-12 23:48:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8b14adddfb949c21af65a85654c543fbebeaa26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 00:06:13.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d054bdea4f0c4c8ab502b4b9e0c29ce0 : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":5},{"id":"e111f189c5a45beb1de979ce0a33c7a9207ac6a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 01:08:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2071bc7e9ff14d4597379f1b467cd5e0 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd26cab26e2d406397ec1b0a494b9cd4 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d463639318f4d80bb71d9d86c2b4bd1 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4e86221a7a149cb9828525d2572bede : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ca51168541c4d3495844df4f7b79abf : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61800bc034e6448590dd145d7396be7b : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/497085fa0a1c4201b8943f5f16a7c68e : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/821b44cdb7324c0c8b87663d256044c0 : SUCCESS in 10m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/801faef31ac140cab94cd57be8490483 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e055eeeb9e144523a880705762d59239 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cc62ead5af92455386cac96eee61e352 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b2b1c02ca3d4288affc485059685748 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c13f6e8890941f9a17185da7020862b : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aa26565fce5048c6b0ccf1cba546c4a3 : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f146b0f79af40919d35842de54a802b : SUCCESS in 10m 48s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e842d83b9cb747d7a862e17c1e0a35fc : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69d1515aad4944e794e96eadd89aceb4 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/82fe3dbf7a324ae397c68759776b883c : SUCCESS in 12m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9251b7b7c1b4b43b6b966dd99babfc0 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f09d7c3f606b4f02b3160a8f8a0b0883 : SUCCESS in 16m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f15c1aa3eed0440e8a20de2d64529f7f : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54541e2d7a0b49719f9f5816de5e6d12 : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f25d6f4ad5fc493bb4ef0919a83a26f3 : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f62a547d1804304ac0d38109a16514f : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/329275e9bfda4342a4f5f287a1856eb9 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":5},{"id":"4410e80cf8f7445f27d193c9ec526e1e5eecfa80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-13 01:45:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b9738a5cd7590494199447854e7c862d46c66d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 01:46:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f0d3a20fabeb677ba879bd156b89fae2af99e6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:35:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a912b8926644b18b8e6d167ead49b0d : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbefe4dfda0b4ad0a32bb2710b1ff6d7 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54352fd1ce174471916b99a2dbae90f5 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/358c77be954549d787856f2519952f8b : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91fc872a07a0430bbb086b7fd938992e : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c8849ed24024808b1a135a62a74c860 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/11e17116019a44d98501e4751881c3ec : SUCCESS in 10m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7dc17e91eada481192e02365e55bad30 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a00495c094a4bd2bce479c8c3fa75db : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a8972626602740629faf48c9b82a7a4b : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8fe889cf17394755a999c05ccd9106db : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21f439dd302f4511ae8af79af3bbd1b6 : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ca686e6b11e4c7dbb172e03d1bc5554 : SUCCESS in 11m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b16a1f423b55402fbfa77b62e5c43195 : SUCCESS in 13m 39s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ba0ca12029d94434930415c114d06da8 : SUCCESS in 5m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/76c492d3638e44529163be4a5d319576 : SUCCESS in 37m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/73fb9b639e2b4e22a603b9212ff3f2b1 : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e95f90aff2a43328cbedebfe348b388 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3904e628953647a59e368f9918e27361 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c8a1c71d57b4f988b11dfbd66618068 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38970fd7a4ba419ea5e8c18efcfb5f0a : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d79a4578f81468f9fccea3e9c30e1bb : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c37b9276b554e48a314abf43eeebda4 : SUCCESS in 51m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c997181449846368623af85f2ee0708 : SUCCESS in 1h 40m 53s","accounts_in_message":[],"_revision_number":5},{"id":"aa76c2f7a37bba76388cc8448f625bc0b64c6b3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:35:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"814e5e0bae863af991a554cf45cad0400455dacc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:37:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107a100505994f0983bb5f436ff5670b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"611b082c85984f55acbcea2add8e78020e089d89","revisions":{"b316a7e57bddae2ee793d0d0aba7110839d7761f":{"kind":"REWORK","_number":1,"created":"2021-04-10 06:22:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/785745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/785745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/785745/1"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-10 06:21:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-10 06:21:59.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id36985024db81eb9c9d6c5f3d8c7b10243029961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b316a7e57bddae2ee793d0d0aba7110839d7761f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b316a7e57bddae2ee793d0d0aba7110839d7761f"}]},"branch":"refs/heads/master"},"a22ff204eabec4c335587f54ea31787e9717e734":{"kind":"REWORK","_number":2,"created":"2021-04-11 06:39:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/785745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/785745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/785745/2"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-11 06:38:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-11 06:38:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id36985024db81eb9c9d6c5f3d8c7b10243029961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a22ff204eabec4c335587f54ea31787e9717e734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a22ff204eabec4c335587f54ea31787e9717e734"}]},"branch":"refs/heads/master"},"b523fcd7c5ad4cd54dbe090204f3d61a79d19cec":{"kind":"REWORK","_number":3,"created":"2021-04-12 06:26:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/785745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/785745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/785745/3"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-12 06:26:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-12 06:26:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id36985024db81eb9c9d6c5f3d8c7b10243029961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b523fcd7c5ad4cd54dbe090204f3d61a79d19cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b523fcd7c5ad4cd54dbe090204f3d61a79d19cec"}]},"branch":"refs/heads/master"},"292cfeb31d5a1f1aef1e29c0768ac61a0e10fd65":{"kind":"REWORK","_number":4,"created":"2021-04-12 21:29:42.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/785745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/785745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/785745/4"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-12 06:26:41.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-12 21:29:24.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id36985024db81eb9c9d6c5f3d8c7b10243029961\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/292cfeb31d5a1f1aef1e29c0768ac61a0e10fd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/292cfeb31d5a1f1aef1e29c0768ac61a0e10fd65"}]},"branch":"refs/heads/master"},"611b082c85984f55acbcea2add8e78020e089d89":{"kind":"REWORK","_number":5,"created":"2021-04-12 23:48:19.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/785745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/785745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/785745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/785745/5"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-12 06:26:41.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-12 23:48:09.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id36985024db81eb9c9d6c5f3d8c7b10243029961\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
