)]}'
{"id":"openstack%2Frequirements~786007","triplet_id":"openstack%2Frequirements~master~Ic9cd9583bf85976446478207ebb2a45d261f1826","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic9cd9583bf85976446478207ebb2a45d261f1826","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-13 06:29:34.000000000","updated":"2021-04-15 04:08:52.000000000","submitted":"2021-04-15 04:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786007-openstack/requirements/constraints/noclob","meta_rev_id":"142596ca407b2a0984b72e466ad81163ace50971","_number":786007,"virtual_id_number":786007,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 04:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-14 20:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-14 20:52:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 06:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 09:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 17:09:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ae5c5a9ee2e257801baae4a4741af625eb5556a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-13 06:29:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b458f7ed277adf6fa67bb4c1e108251cc4214074","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:42:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4b30d42536b91912fdcfb3c37ed6fce658e38e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 09:34:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300d278c38b84587a061b9751337a121 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e2d21cc1bd45a78b9a1fbd1be24fc7 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57a2978c2cfa40c494c6186c31466bee : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e658c2597f74adf9b05301828d807d5 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/115554b2737b453e9a8e2b718562ffe0 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0e0570abc884ab2973fd00c27751c45 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42c5e9ff3e2e45efbc66494eda507116 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/493a66bb90644904935ed933ca1d9867 : FAILURE in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/723abc8b657c49a6917d17d4aa7d6e86 : FAILURE in 5m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7309decade9146b8a9253393767491fd : FAILURE in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a75ced31595c43cc89495906e69b6222 : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e43ea3f3be7e4298b34e0f71b2708870 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3cf79267d4541a9b19a478c74511393 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7483307a3f9d4f24b41a0cd074ef8d1e : FAILURE in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6bcd59484eb40e1bc572b27f815dc24 : FAILURE in 10m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/71fd5dddd5814cc38b6360ea31e5740a : FAILURE in 4m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25277559a9df4dd8b32ec8dcc6d28e74 : FAILURE in 31m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d7619b74089b464bba730ba33da8319e : FAILURE in 11m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23b4e804779d487d8a9046388345309a : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d643734c39a141c4a5d51dedbe2b0091 : FAILURE in 22m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8b7b6f34397406d92ff2f0c86ea74cd : FAILURE in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a50ad4b635b0482db3f14f983705eb53 : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5ad79c1ed6e4bad8d64e1e1de271571 : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7f91db9100e46ada3177cf66d5a6743 : FAILURE in 2h 19m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14654ea86f9a48da9516bd39c4b1f3dc : TIMED_OUT in 2h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8590b6999c871da995715f9a8b78e70df2d9791d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-14 06:28:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5dd9545ee6a0a3d155ef791da4b77c9413b4035","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 06:44:50.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/93d59686d2ee46fdb58509dcf3db34ca : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ed520caa1b71d4956c3f457b8090f113dc27eb73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:07:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d94170db2546db9a76e5793b9d7add : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf60de424bd43f68be589d51d30361b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a199eac4d41740058f47464b1e152a52 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b5451a7dff84e7fb8e32a9c27fea4b4 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d85922e2bce4a24bb38c7583461640e : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6cfb48ef960496189c03f4067ce300a : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2567d7df1bcb4a39a890059f9a9b934f : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7cfcb29c2b3741618575c7670f417f2d : FAILURE in 7m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a054f28a00d4255b8b3f2cf2dabb144 : FAILURE in 6m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c98802cc78cf49098490f06904665c04 : FAILURE in 10m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3fb49d4f2fd844be9d0bc60d887277d5 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc4e01f7952e439d925e7535c1ce0106 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef69bbd823424d08833851f3e920c8c4 : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69972c5829f04784a8d328beaf3fa720 : FAILURE in 9m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1884f3631ea745cbb1eadbf6d36b0ca1 : FAILURE in 11m 19s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1ec566ef7c9d4e088a903dd1e39c1289 : FAILURE in 5m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/56f82de4871d40a9a50079014e15b245 : FAILURE in 26m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9904724065a84259805214f52b947db5 : FAILURE in 12m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fce469af0df345e8aac11aec9bf56e8a : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b5ca1d4f1c74f9ebc8c66850c5afdfb : FAILURE in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77cf3a3239e149828189dee4591e1973 : FAILURE in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff49cea3deda4d438fe837911901bc18 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9881d22abc034630874007dabfc9f333 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd26d848d03c46688fcb3297499bee78 : FAILURE in 2h 08m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d6de4c681054a368bf1b57298b10964 : TIMED_OUT in 2h 03m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d6199eb5d240451d2d94ddaa7dc58097f8482eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 17:09:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa90dc2b00c5b92d9e145473dad2b54411947d6c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 17:10:56.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"d28b23cdcb2f6120bf124b844df4b1b2571eed37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:42:55.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ccdbd7fd8823410fa261c59a63593f2f : FAILURE in 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"8a66fa3d7f1a9aad34fe16abd6711f212a4931c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:01:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff41ece736c34ab099c25ad91b4a34af : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4fef3abe114e6aad485fd859035f34 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b40894304e4f70bb89601398021f76 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dd11b462b1a4e51abc54d49677756c8 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60305b1689504f298248f30ac3a16dfc : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8257813c16e429bb659721ce1d0a0af : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/953145c2b1244e1483caf884e83ded1b : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cad4a6c6d4db460da7e0cd2291b1ae99 : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3bfcc5573bdf4d688a46860585a3b5de : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90f8a96f1fdb4370a02f5ec71756f6cb : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/060fe9744518417586bd12e8a2edaa50 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c67aaab433414a0e99910d0442e59579 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1f6a82ad3b145fa83f82282d4934055 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/332de85bc8d646638445f9c7a1144fa1 : SUCCESS in 12m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a9c1ed71a41495cb9ceb32a6ecab8a7 : SUCCESS in 15m 43s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3ea27a2f99694eee8c3e79a8e4a94cc6 : SUCCESS in 6m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c47852763384eb8ad92488dbb5683a9 : SUCCESS in 30m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d0feb8de458473da3c00cf6e853de5e : SUCCESS in 13m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7cdb46a5fbd479689aaefb7b1b6515c : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c87c313beac24a61973f2933a7a75c11 : POST_FAILURE in 30m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eca4166a95c042968d9a0a2405901a79 : SUCCESS in 8m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7573f6968ca4ac9b125bea119a1d9ff : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66e19a34234845f0bf1555d634865ec2 : SUCCESS in 14m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ed08214f25c4103a2d55d97ae3dfa4e : SUCCESS in 54m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f380813bd0d045d4852c20dfccb1db3f : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":3},{"id":"41c242caa7ba04141b432f60995bc5f8495e7a0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 19:02:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c47049c73ce4fdec87582ff6aefaf75a171bb89d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:29:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8aa2c05179bb449781c33545129bc2ae : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a6e44acd0c600cbbfb1664b69218b24a96637f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 20:36:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bff53be2b0c44b79d5f33f1bdc7789a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c194531cf34728ad12bacb807e00ef : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/341e3fc9ff27449aa9633b8021cf0255 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b27f65178894e6e9da9031a4fdbf786 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f74a4ef74fa40878e313b3765ed290c : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9985df03d3534034b52797f4f7839b0b : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52f1a34cc0a4471a9cc5ce29274373d6 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31ae2637711f4c2c984182e5723f008f : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af3fce266e0644d9959e8cce0561333b : SUCCESS in 15m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d4762350c5947e6af407871623913e5 : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b4869374d7a144eabc3edb96d88a5fe1 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc78b8030e864817ac7edc677b1490ae : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/159ad71ba6444699acc5b8b50da951f1 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc60fe0546e94ac289c29de17c9b3570 : SUCCESS in 13m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cef21f4945d84cd1b78aa134910cd38d : SUCCESS in 17m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b7b3de0e934b4b2aaa8641f3a20977e3 : SUCCESS in 6m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b523b113d07a4b9b9d3eed501570bd6e : SUCCESS in 30m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cdf504ff6794afbb197449e34d764e9 : SUCCESS in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85a18c8bbdf84c008961d0f480358e84 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60c9c4467c9448e4ae23c976059d5ce4 : SUCCESS in 21m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5bb73521b0844d38807a1a38fe4a16c6 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a552c4165924d40b987eaf4718281e5 : SUCCESS in 7m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9892a93d00f0460a9c958794a5e11ed4 : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9965ddad8b364520b8f387a5e5630fb4 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f81acb233eb1489482b0f1ebf73e82ae : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":3},{"id":"24a7736f9ec8d77073f58a8b7f08aeef28416f6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 20:52:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7697ae754a0544957c85b4e9fda750ec5fcf96bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:08:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58d4d606f7efea330e04a6e479b61e0abb63fe89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 04:07:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0f9daf53d6450d8d8a552814a74d42 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8aceec6224477c8135516e9cb96b37 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f68d9743e764ccf92998fadc197336c : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1d7f86ba97496daaf164af20a859bb : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bc7c2660dab48d7889413ad7b001367 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75ae8c1c39a14494973a84ad2adf622b : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4755ea10cc1741179b9ddcd87cfb92e6 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd3663b84bed4aab92de8b1f73f6eafb : SUCCESS in 16m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41aed667515c45c2833cef6b60414eb0 : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/58cc2a7e2f954fa1838f853047730cb5 : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1b77396f25ba47f0a3a4718130b38fff : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d56af6eb99547a9bcff82798d4c706c : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0b9fc9ade6b64325be366d1d3b32380f : SUCCESS in 13m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/275a8217328a4a91a46876f0e16d46ab : SUCCESS in 12m 44s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/81a06300299f46b298479c2d49ed07e1 : SUCCESS in 5m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/33f0afb21b0c4155890bf29aa6acb02c : SUCCESS in 23m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68db563a4b4642d8a80e481be0d508f8 : SUCCESS in 12m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd695401a6bb4e07803ac3a58d5b37ef : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/942d2f85528247e79458604c9c3e4dc2 : SUCCESS in 19m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf36b7e5ec4c4618a1759a4e917cc1b7 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d01690dbfd74fcb872862879e94b573 : SUCCESS in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f714315e5cd74b48950e27e38532e9fe : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca309d02bd2c4860863dff2c82fa6cea : SUCCESS in 58m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c16c8b589ff4f20bc3306af0661c59d : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":3},{"id":"a7140236a1458cc9202e8ad5da970154c0995edb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 04:07:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"142596ca407b2a0984b72e466ad81163ace50971","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 04:08:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2deaf606ff4ca0bbcf35fbac007b20 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d","revisions":{"43ae37f6a20941de17c0676eb8c1eb70ca86792a":{"kind":"REWORK","_number":1,"created":"2021-04-13 06:29:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/786007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/786007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/786007/1"}}},"commit":{"parents":[{"commit":"1e4b80367f8f2e90b08d3d87b3f80e7da338c215","subject":"Merge \"Add pycdlib to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4b80367f8f2e90b08d3d87b3f80e7da338c215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-13 06:29:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-13 06:29:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic9cd9583bf85976446478207ebb2a45d261f1826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43ae37f6a20941de17c0676eb8c1eb70ca86792a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43ae37f6a20941de17c0676eb8c1eb70ca86792a"}]},"branch":"refs/heads/master"},"7c57003c1b3b5bf15f2a05cfeaea8045d556f664":{"kind":"REWORK","_number":2,"created":"2021-04-14 06:28:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/786007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/786007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/786007/2"}}},"commit":{"parents":[{"commit":"611b082c85984f55acbcea2add8e78020e089d89","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 06:28:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 06:28:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic9cd9583bf85976446478207ebb2a45d261f1826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c57003c1b3b5bf15f2a05cfeaea8045d556f664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c57003c1b3b5bf15f2a05cfeaea8045d556f664"}]},"branch":"refs/heads/master"},"bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d":{"kind":"REWORK","_number":3,"created":"2021-04-14 17:09:49.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/07/786007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/786007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/786007/3"}}},"commit":{"parents":[{"commit":"611b082c85984f55acbcea2add8e78020e089d89","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 06:28:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-14 17:08:36.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back tenacity\u003d\u003d\u003d7.0.0 and sqlalchemy\u003d\u003d\u003d1.4.7\n\nremoved pkg-resources and re-added setuptools\u003d\u003d\u003d56.0.0\n\nChange-Id: Ic9cd9583bf85976446478207ebb2a45d261f1826\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
