)]}'
{"id":"openstack%2Frequirements~786207","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I44b2532bdbe07a9725150b355b6f5c5a9c14fb04","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I44b2532bdbe07a9725150b355b6f5c5a9c14fb04","subject":"update constraint for ceilometer to new release 16.0.0","status":"MERGED","created":"2021-04-14 11:18:58.000000000","updated":"2021-04-14 23:17:14.000000000","submitted":"2021-04-14 23:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786207-new-release","meta_rev_id":"d83f44b4ea3a7ba3835e8bdd7af49bdc8d562e1e","_number":786207,"virtual_id_number":786207,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 23:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-14 15:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-04-14 11:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-14 15:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 11:24:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-14 11:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 12:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 15:38:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0959581c1c4e8a9e368e7edccf1fdb0ccd6394f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-14 11:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa4c9304030162cbb1c3a15b14c88136808642d2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-14 11:24:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4bb5ab7d0b803b2a08d6a7e9611d1d7acf8bda47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:34:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5d0e63bfd32f403da53bdf43f07a8cc5 : FAILURE in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f19f867ad884745a59f1e27963f0298db703311a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9848907ca9ef4ecfb3b739311e3c7fcb : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc7a3f2eeb64e7cb9bd5b5d990a8925 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d90d5bb8fac4ea4a257c9c6395d50bf : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8decd2db3e60418cae8de6ddc5c1af03 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30f64a79d4d84a639e450b8cf3a7a51a : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57075c0e34ca4141b94ec12ac633fdf7 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f67db60198244804bf15dc0925c3aabf : SUCCESS in 8m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c40a98f4ad5141d4be35240d6e0dffa7 : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0e5e6561c47048ff913599512dfea429 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce7d84e51ce84c479ccc23db770d4512 : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f83822c23f044a73b592c8c38e786f21 : SUCCESS in 13m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/494d7d3092dd4571b7f5ceabb8fd0006 : SUCCESS in 7m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86df9976387944858d7f78a49fe5e44c : SUCCESS in 10m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eec494f0d9df410799c0e540c789d80e : SUCCESS in 13m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8746449d10954fc5b3756ae2c52e071e : SUCCESS in 6m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d499ecba20464ad387b97d6a92536a73 : SUCCESS in 29m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28a08133dee243c785d508191f3474a0 : SUCCESS in 14m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b496cc671f4a44b4899a031f19a1b1d6 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e0e276b5a7c40a29998119a891092fc : SUCCESS in 20m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e35434fc58e049de855ac0576c5f2a8d : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ab6ae8ace85469a8b41147e0ba76bc4 : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24b27c267cd74cf2853910c8d0cd426c : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8768edd289647fcb1479505d8bef4c5 : SUCCESS in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6d5e65c7b4540268dbf801629d7d64b : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":1},{"id":"90ce47d9420c35311284063859b7d27ba2633a0d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 15:38:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1979aaa996c9b6308aa2fb4e785b901eeb6c6b72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:38:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d1ab7e3f60446e1ec0b1e2366a3c96ac29e073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 23:17:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42334060579b4ecc9926f32aba082ab5 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f64ade0cf5e4ae0aca06ba93bb76b9d : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b52f6661ce14ea39d7c2182bb19018a : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ac2c79f52a9455ba5ad1376b649b0fd : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fbd74f00332b4694a67658232c9f6379 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/363dbc0b24174f7b9c98d7bf35f3c211 : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3dd650ce1d484962944a74d74e3b9ab0 : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c76a5d4d185d42e8bbff6d3161eed945 : SUCCESS in 10m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7164afcc8e6b43879c58a5b7ad66e58a : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/94b05e511c894d69b5e25b3531880977 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1eadbdd6f4374df8adc8a8c5cd817555 : SUCCESS in 7m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a09ff3b3fea2441bb2116915470ff71f : SUCCESS in 11m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f497b632d1644c2986c07011f75dc9d : SUCCESS in 14m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/699f1131c5c449d8bab3286877642204 : SUCCESS in 7m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ec9626e307a49fa831f8a8081beb7cd : SUCCESS in 25m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/195d241141834e8abf64d24487a84289 : SUCCESS in 14m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e534fc4abb554b979574aad6572c1a31 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8521d2284ef34e1db856aa01a5338818 : SUCCESS in 16m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9afdf1d830184956b3c7ecaabb9565d4 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/082c1fa0712b4452a87499ccc591fde6 : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e5b90e52fd484773ac665b7379241b54 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0108742dab0b4fe58a484eb955949947 : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22c9fd66edd0446e8fa55b5e516f9c66 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d83f44b4ea3a7ba3835e8bdd7af49bdc8d562e1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 23:17:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"686cd95bfb865a202d78a4ed72ddc615a56265db","revisions":{"686cd95bfb865a202d78a4ed72ddc615a56265db":{"kind":"REWORK","_number":1,"created":"2021-04-14 11:18:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/786207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/786207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/786207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/786207/1"}}},"commit":{"parents":[{"commit":"17b9c21a43c0cc40e58d10416aa686c417673f8e","subject":"update constraint for horizon to new release 19.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17b9c21a43c0cc40e58d10416aa686c417673f8e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 11:18:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 11:18:56.000000000","tz":0},"subject":"update constraint for ceilometer to new release 16.0.0","message":"update constraint for ceilometer to new release 16.0.0\n\nmeta: version: 16.0.0\nmeta: diff-start: 15.0.0\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ia6d79df224a792915976864857fe2271e031b54d\nmeta: release:Code-Review+1: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Code-Review+1: Andrey Pavlov \u003candrey.mp@gmail.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\nmeta: release:Code-Review+1: Carl caihui \u003ccai.hui@zte.com.cn\u003e\nmeta: release:Code-Review+1: Rafael Weingartner \u003crafael@apache.org\u003e\nmeta: release:Code-Review+1: Xinran WANG \u003cxin-ran.wang@intel.com\u003e\nmeta: release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta: release:Code-Review+1: 刘雪峰 \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta: release:Code-Review+1: Martin Chacon Piza \u003cmartin@chaconpiza.com\u003e\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+1: Douglas Mendizábal \u003cdmendiza@redhat.com\u003e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+1: Maysa de Macedo Souza \u003cmaysa.macedo95@gmail.com\u003e\nmeta: release:Code-Review+1: Adrian Turjak \u003cadriant@catalystcloud.nz\u003e\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I44b2532bdbe07a9725150b355b6f5c5a9c14fb04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/686cd95bfb865a202d78a4ed72ddc615a56265db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/686cd95bfb865a202d78a4ed72ddc615a56265db"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
