)]}'
{"id":"openstack%2Frequirements~786254","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I76ae1895b50e15b243ccc396fa10527a425901b3","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I76ae1895b50e15b243ccc396fa10527a425901b3","subject":"update constraint for os-net-config to new release 12.3.4","status":"MERGED","created":"2021-04-14 14:57:12.000000000","updated":"2021-04-15 02:03:03.000000000","submitted":"2021-04-15 02:03:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786254-new-release","meta_rev_id":"5fdf7bf8beba3ebcf9ab2e2a34db1f3befdb39fe","_number":786254,"virtual_id_number":786254,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 02:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-14 17:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-14 17:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 16:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 17:07:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b76c0c0b0c579841e2ea2ce04e2add9d76c85797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-14 14:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3ff535f4de90fbf767c6a55d96874948e34eae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7813fe5725cf4be49f89df16b3a85c7e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3783831138964963bedc155db71d19a7 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f184bdf986564cd5a254d9c99a4bd520 : SUCCESS in 10m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb01471e5b424df8bdf5a141769963dc : SUCCESS in 8m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54c26f811d644d9b8bba97c9bd65630a : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2db80ac24d5a46f7948410f9cf96b5c5 : SUCCESS in 10m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/38453a7c07e64f27b50c396cf5b28f12 : SUCCESS in 17m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b01f0bbb2774037a648d4db5060d32d : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7a1167318444f4f8f474019e5cd014d : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0c0a3357b384497b5d1cf869a3f39be : SUCCESS in 16m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee405f9a567b413baa09f4c130401493 : SUCCESS in 25m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c0dfbe316a8f43cebfc011339f24f4ca : SUCCESS in 25m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c60ada14e0d4a87a264280772f298a9 : SUCCESS in 17m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d48da14a310444586cffdc94676e749 : SUCCESS in 19m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e5298ddefb445e6962585f058d9fbaf : SUCCESS in 19m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/abcb1ba9eee047e49d01228e9f3a6911 : SUCCESS in 7m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4c38aa4e78e7481daa6ca8b2745097d4 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45e4a93662354dd1873efc940c728865 : SUCCESS in 11m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d685e82f30e74ca18a739f9686af26a6 : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/557c11a24353451c8e6de814f554cfb2 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe366b9b1f974e51af3ed47981847dd8 : SUCCESS in 55m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c070a73c0a347aa95ebd2bf3d2e88e0 : SUCCESS in 1h 13m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/720d34a22edf414b97b18328adacacba : SUCCESS in 1h 29m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/302b9e2182a54fd68c81c5f76386eef2 : SUCCESS in 1h 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3dec19e45a14bed6e608fd398b83f5d1ad10201f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 17:07:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d875a7e0351e38773ea74c06b646e7e72f64af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:07:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b505d3d3a11c07b0d02ac0c34f14118d4994ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 02:03:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ea345405014d2992ed9de51c2dad2a : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77054029c3324aa4b81a77c0146131b6 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1e6e91682d046238dbbf7ae8f516308 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cca57ccb7d5040e2a2c18d5796f310fc : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1e1955e757d4ffd8f95c323803142f3 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4501121369e49bca58db17b257c2333 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cfa76f499a7a4096b8f409f2fc85cf80 : SUCCESS in 11m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a85b86350e5f4fcfb5f244bc8401dbe8 : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a849e7109ad484283330f795b56d57f : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55017a3a89ee4ed4b52cde74494e72f0 : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba5ce3ecc1084c548834b2eaaa1df433 : SUCCESS in 14m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72502bb23814467ea6ba76974af18ced : SUCCESS in 26m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/724b1218acc74108af380104df82068a : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e6fd248e61c64a35ac64b8e90ba8cc7a : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e91130d0032c4c5e9e29cc7dcc88be35 : SUCCESS in 22m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/41e24ed22a4f433cbc03551e95391160 : SUCCESS in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2e45d63b5831492d9cc3fc21496f893d : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3780dd4252b44464998194e7f9f78c34 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d97d4aae7d6540f288fd6000c3512a70 : SUCCESS in 12m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f219d554161e44d1a8dc49e4200f046e : SUCCESS in 8m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf16f8c3562f4297986cd393d6a6a3a8 : SUCCESS in 59m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3452f9804994ec0975af34d6dcc90cf : SUCCESS in 1h 20m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/607f787b63e34614a20be0ed45fab4b4 : SUCCESS in 1h 27m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/ce1c8e0625c64fd7bc686eebce74094c : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5fdf7bf8beba3ebcf9ab2e2a34db1f3befdb39fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 02:03:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"850d56dbaef898fbdc16262f3440293da73096e3","revisions":{"850d56dbaef898fbdc16262f3440293da73096e3":{"kind":"REWORK","_number":1,"created":"2021-04-14 14:57:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/786254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/786254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/786254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/786254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/786254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/786254/1"}}},"commit":{"parents":[{"commit":"db35b054cc0aca4055c6bdd7cc1d5cca3a4a687f","subject":"update constraint for os-brick to new release 3.0.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db35b054cc0aca4055c6bdd7cc1d5cca3a4a687f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 14:57:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 14:57:09.000000000","tz":0},"subject":"update constraint for os-net-config to new release 12.3.4","message":"update constraint for os-net-config to new release 12.3.4\n\nmeta: version: 12.3.4\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ib5dbe93c16343a5192e70e0577c500f21258c126\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I76ae1895b50e15b243ccc396fa10527a425901b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/850d56dbaef898fbdc16262f3440293da73096e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/850d56dbaef898fbdc16262f3440293da73096e3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
