)]}'
{"id":"openstack%2Frequirements~786256","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I688718967cb9be717c58d9f465d34fac49282e49","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I688718967cb9be717c58d9f465d34fac49282e49","subject":"update constraint for paunch to new release 7.0.4","status":"MERGED","created":"2021-04-14 14:58:11.000000000","updated":"2021-04-15 03:56:50.000000000","submitted":"2021-04-15 03:56:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786256-new-release","meta_rev_id":"60f69fbdd32964e5a60a8d562a2a95f1b5a9b372","_number":786256,"virtual_id_number":786256,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 03:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-14 17:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-14 17:16:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 17:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 17:16:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"49de3722acb510e9c4476de418b84454f3a9ff37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-14 14:58:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74529c27cf2be5e769e32e2378cfda7b22f2eca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:13:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c213c690bf3a4efeb2e2c1065f7b05e4 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d74569b882de4702adef92578e775480 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ad5c349d4014c12bdd8ae8aa7e86def : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2921e88503714145a36a651e3c8ed5c7 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f662837b3bcd45a18e6ade55787aaaeb : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5eac0d9a06bb45bcbd7a3d4360a74bc4 : SUCCESS in 12m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6df38086e84649db951b2556c313421f : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e06d7adb4e64ff28589084f0b6cfaac : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/604d095a6b01401c9bd9316c9087641a : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf75c9df8bb5419db72ce793fe96b20f : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/573eaa8ce7c346f6821e79cf344b738d : SUCCESS in 19m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8505b78d45f045e6afdcacc34a78e77a : SUCCESS in 39m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42b0f303f61a48a2aad9344b1bb52d9d : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2caf67026fd24077b32bf2b11702e450 : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc73d0cf60b14813a80b161f60354bb3 : SUCCESS in 16m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8095a2e0091413088dc123cc492555a : SUCCESS in 5m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2a1301103bdf48d1a8a0413f99f81651 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67fb5b627e8347839f98501e3470cfee : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6a06299c15b64554aa0ac0d68d7f9a79 : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ad6e6f6a2fd640d39e13b7319ba7c00d : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c90cc202ca74dddae9605c94ab02a0c : SUCCESS in 51m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c20a9cf3b4c4eda992f5b8b8beace15 : SUCCESS in 1h 14m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87e7f46e2e1c409b877d6398db294bd2 : SUCCESS in 1h 30m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/fb02797ed71e470b97b6c680a782cdb7 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"546e2161f5c2be459503d7146920640d44166c65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-14 17:16:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c20719061fcd47b42ef4d65a6c1ee1461a12e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:16:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e214de37ef3abe42667eed8d1afc75f282d59a9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 03:56:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075d6eba46e648b5b08673ac2737e892 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ce5a7431f494b93907e659dc9e4f13c : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/720fcc044b8b49fcac6e6b1bc719a5d8 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/195bfb887416489b8d52743c293e2128 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55a6655b8f8b40c99873a8a1c5f91809 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71e591cb851d493ba3d5266e32bf39e1 : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5fef20a2a6e0463e863f102f307fbdbd : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec7f7eb38eff448f96cb20187e192d07 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fe431b58e42418abbd4796df315b105 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87fa7eb5a5ce4ff2874592f4fe98c7a5 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/637c19fd92744e408d3cb614336b53be : SUCCESS in 14m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da00f4ce3fb249ff93bb5e307c0718c0 : SUCCESS in 36m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f30f5b2e1d3a4baeade6077fc7475c0d : SUCCESS in 15m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d2da033899f54bb78f5072fdea142c48 : SUCCESS in 14m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d98945c93602480aa5fc7c3a328f514b : SUCCESS in 24m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d722f21e09604309a530629fb941964f : SUCCESS in 5m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1a70708f925e4f0dbedb2c6cdf3d49ad : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/034a488f52d3436c931344187a982fbb : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a77de51cfe664aa0a545a99eed9f9648 : SUCCESS in 11m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/80196b4810f1497ca79f4483230781b5 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3193c699bb834803b8fc3f65ee954d92 : SUCCESS in 54m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3cb9c5792444ec49a35fc5fd4b57c52 : SUCCESS in 1h 26m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f44a39c3bbe47d2bcceb09839efff74 : SUCCESS in 1h 24m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/27086392f84c410ebd60e26fd00d0a58 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"60f69fbdd32964e5a60a8d562a2a95f1b5a9b372","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 03:56:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fc40bd02ac94d46f663fb0157d1bc014393155d","revisions":{"2fc40bd02ac94d46f663fb0157d1bc014393155d":{"kind":"REWORK","_number":1,"created":"2021-04-14 14:58:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/786256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/786256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/786256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/786256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/786256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/786256/1"}}},"commit":{"parents":[{"commit":"db35b054cc0aca4055c6bdd7cc1d5cca3a4a687f","subject":"update constraint for os-brick to new release 3.0.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db35b054cc0aca4055c6bdd7cc1d5cca3a4a687f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 14:58:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 14:58:04.000000000","tz":0},"subject":"update constraint for paunch to new release 7.0.4","message":"update constraint for paunch to new release 7.0.4\n\nmeta: version: 7.0.4\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ib5dbe93c16343a5192e70e0577c500f21258c126\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I688718967cb9be717c58d9f465d34fac49282e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc40bd02ac94d46f663fb0157d1bc014393155d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc40bd02ac94d46f663fb0157d1bc014393155d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
