)]}'
{"id":"openstack%2Frequirements~786338","triplet_id":"openstack%2Frequirements~master~Ib3eff93fc7f793f7c442904247c70898501c0570","project":"openstack/requirements","branch":"master","topic":"s3","hashtags":[],"change_id":"Ib3eff93fc7f793f7c442904247c70898501c0570","subject":"Add s3transfer","status":"ABANDONED","created":"2021-04-14 22:25:53.000000000","updated":"2021-04-20 06:56:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7e40926d15248254a9240f49df85c401bcec156","_number":786338,"virtual_id_number":786338,"owner":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-19 22:25:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 23:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 19:58:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"dbdfebfe139a47381912e0de7057e2727918feda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"date":"2021-04-14 22:25:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9f0451b377e3cdf291e77192395cf32224aa660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 23:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d409aa43c84265b1e4cec35c500e98 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f58303ed62400a854a5cb704268da8 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12652fb3e7dd426d9867203fb06ca97d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db53e3c5f2c54cdaba8cfb1b6c354e50 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f92907964d4453e9212b2cc7bc1dc0c : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d28091d961fb402eb0806c238ab71084 : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61518e1de9ff4149978e939e74ea040d : SUCCESS in 43m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dc2f78d8cf44ab5b22daa3d2e339129 : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6869ecfcd0d0c32ba3a38805dfd2dc768c615e3e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-15 19:58:06.000000000","message":"Patch Set 1:\n\ndo you mind answer the qestions from the following link in the commit message?\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":1},{"id":"1bd8eecc5d32794335bea892d8f643636b0b9f0f","author":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"date":"2021-04-19 09:10:06.000000000","message":"Patch Set 1:\n\n\u003e do you mind answer the qestions \n\nnot a problem but on second thought: s3transfer is a dependency of boto3, and boto3 is already there and runs fine. If we are not adding any explicit dependency onto s3transfer, the requirement check of the CI pipeline will just run fine.\n\nWhat do you prefer?\n- option 1 where we go ahead with this change with the right commit message ?\n- option 2 where we just discard this change, s3transfer will be installed \"silently\" as it is  done today for components that depend on boto3","accounts_in_message":[],"_revision_number":1},{"id":"293804983e66eb70b57a79f62f9eb52a826bded8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-19 14:23:30.000000000","message":"Patch Set 1:\n\nI\u0027d prefer you update the commit message.\n\nIf you have a \u0027import s3transfer\u0027 line in your stuff then we should track the requirement here.","accounts_in_message":[],"_revision_number":1},{"id":"21ba9d244d55fbbe7620329e52e1911e9313bd65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"date":"2021-04-19 21:04:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b729efb65058eb70864aa5045f181e6545915898","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-19 21:12:25.000000000","message":"Patch Set 2:\n\nI don\u0027t think your story / review actually has the `import s3transfer` line anywhere.  If you are using boto, which calls s3transfer internally then boto alone is good.  We only care about first level dependencies (not dependencies of dependencies).\n\nhttps://review.opendev.org/c/openstack/ironic/+/784625","accounts_in_message":[],"_revision_number":2},{"id":"51d4ab892fae788ff9d05b3aaa3092e8e53f5b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 22:25:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7add3a5140450d9a26f91db4dadc52 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7984e7cf8c9446b8da0381dbcc0521f : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ee4fd6cfd64a4a93868af2503652e9 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4db372f4069f4a22a3dfecdfe0697aac : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29b18d1b751b4f00857b31ce9fb98ccf : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c03abacc5a148a19e581c82720dc004 : SUCCESS in 4m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/153d966779d3478db9944c9ec897cc28 : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09b661809bc8463ca6385f1a51a78429 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f7e40926d15248254a9240f49df85c401bcec156","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"date":"2021-04-20 06:56:29.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18491a79a2ac6951a8869ed546f4a15d360f6e8c","revisions":{"f54eb380005cc6c51733ceb910c8ada73eb47d73":{"kind":"REWORK","_number":1,"created":"2021-04-14 22:25:53.000000000","uploader":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"ref":"refs/changes/38/786338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/786338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/786338/1"}}},"commit":{"parents":[{"commit":"611b082c85984f55acbcea2add8e78020e089d89","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}]}],"author":{"name":"François Rigault","email":"frigo@amadeus.com","date":"2021-04-14 22:18:48.000000000","tz":120},"committer":{"name":"François Rigault","email":"frigo@amadeus.com","date":"2021-04-14 22:22:02.000000000","tz":120},"subject":"Add s3transfer","message":"Add s3transfer\n\ns3transfer is a dependency of boto3. boto3 is already part of the\nglobal-requirements file, but s3transfer is missing. Both components\nare already part of the upper-constraints file.\n\nStory: 2008671\nTask: 42266\nChange-Id: Ib3eff93fc7f793f7c442904247c70898501c0570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f54eb380005cc6c51733ceb910c8ada73eb47d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f54eb380005cc6c51733ceb910c8ada73eb47d73"}]},"branch":"refs/heads/master"},"18491a79a2ac6951a8869ed546f4a15d360f6e8c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-19 21:04:04.000000000","uploader":{"_account_id":33176,"name":"Francois Rigault","email":"rigault.francois@gmail.com","username":"frigo"},"ref":"refs/changes/38/786338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/786338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/786338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/786338/2"}}},"commit":{"parents":[{"commit":"611b082c85984f55acbcea2add8e78020e089d89","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/611b082c85984f55acbcea2add8e78020e089d89"}]}],"author":{"name":"François Rigault","email":"frigo@amadeus.com","date":"2021-04-14 22:18:48.000000000","tz":120},"committer":{"name":"François Rigault","email":"frigo@amadeus.com","date":"2021-04-19 20:57:25.000000000","tz":120},"subject":"Add s3transfer","message":"Add s3transfer\n\ns3transfer is a dependency of boto3. boto3 is already part of the\nglobal-requirements file, but s3transfer is missing. Both components\nare already part of the upper-constraints file.\n\n* Is the library actively maintained?\n  Yes - latest release today [1].\n\n* Is the library good code?\n  Yes.\n\n* Is the library python 3 compatible?\n  Yes.\n\n* Is the library license compatible?\n  Yes - Apache 2.0.\n\n* Is the library already packaged in the distros we target?\n  Yes [2] [3].\n\n* Is the function of this library already covered by other libraries\n  in global-requirements.txt?\n  No.\n\n* Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes it is a dependency of boto3 already used by Swift, and\n  used for a development on Ironic.\n\n[1] https://pypi.org/project/s3transfer/\n[2] https://packages.ubuntu.com/focal/python3-s3transfer\n[3] https://src.fedoraproject.org/rpms/python-s3transfer\n\nStory: 2008671\nTask: 42266\nChange-Id: Ib3eff93fc7f793f7c442904247c70898501c0570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18491a79a2ac6951a8869ed546f4a15d360f6e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18491a79a2ac6951a8869ed546f4a15d360f6e8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
