)]}'
{"id":"openstack%2Frequirements~786360","triplet_id":"openstack%2Frequirements~master~I1da74bbd42defa7e2b22a9799fdc48ae4cdafa8b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1da74bbd42defa7e2b22a9799fdc48ae4cdafa8b","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-15 06:35:26.000000000","updated":"2021-04-15 20:34:26.000000000","submitted":"2021-04-15 20:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786360-openstack/requirements/constraints/noclob","meta_rev_id":"fed932455caddb3d39072f6a611db90517e692a5","_number":786360,"virtual_id_number":786360,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 20:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-15 18:45:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-15 18:45:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 06:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 08:53:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 17:18:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ae0f31005de43bbbf139355d487fb3a5e2f30400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-15 06:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1fd13c1bbdd542db562098739988a8fd087c272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 06:52:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3af495dfdb774d9b8a66fde2420bbd0d : FAILURE in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"03d678e19d5f69deb1ed3df5967952ca322a7e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 08:53:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69f3dde8f634cb490e545a88786bace : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b063ec3a3084cf195e2a96991bb0a76 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2500a346ade946f3b2c29350f26c7f5e : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71829dcd834e4cf8bf01d66042f3c419 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b30fd62db65487bb83ac4c52dba01bc : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/851e20cba1214d4ca30d0997a9d5417e : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eae2fa70b23a4239aed3b934915010aa : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a47bdb95fa04e9993c9e5e7a8eeb9fb : FAILURE in 5m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/47209e9b38c14c96a9cb2e7ec14f9ee0 : FAILURE in 6m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6ff407da74af4114b61dcc8a5db2019d : FAILURE in 11m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/36907a91981f4ce0accb4528f4f8f989 : SUCCESS in 8m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6dd0862c9ae744798b3765e704302ba7 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71eba8f47d5a48c793dec5bd969f939b : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2586575a2dc84f688311e2468889defc : FAILURE in 11m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53994c3a0efb49188d0d25a573cb2280 : FAILURE in 11m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/57766991f2224b7382553eef930e1a0d : FAILURE in 5m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a9e732fd551d402ab629e8d4cc82a765 : FAILURE in 25m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6066195ca8e94f4faec21f96885d75f7 : FAILURE in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0fa2f675b6349dc88435210e9a05ea5 : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78e0d400940a4dceb4d876988cd33581 : FAILURE in 19m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2b3db72c183c4f79a39857f8e82e4402 : FAILURE in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6846f74a1e54e198b68b36f9edaa313 : SUCCESS in 8m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/124effdcee1847af8e3a0644aed5cd97 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66035f63fe2745359db40f7562a45b5a : FAILURE in 2h 04m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff7dd676636a4ceba6ae876af9bbe167 : FAILURE in 1h 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7083a74a3509289484c7aab024799100bba4f3d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-15 17:18:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba1311408ba2b387fd2f9b2b5e6a849076711134","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-15 17:21:07.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"563390a1e6d1f5e30c56af595b21a6141bf8ea8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 17:53:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0f654e49d7e244f48ace9fd82bf3b32d : FAILURE in 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1a484d5ba5ea444ae1931e5aa03a68a6800b7c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:39:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac65c31b4da145be8ffca766f00dc6de : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3ce88f4755416cb681b0626b11a40e : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71a19d31d6194884b7494c89269fae78 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/372de3898f7e4f268d2418ad940e3dc9 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b66660de9b624edba1dadff262c8cdea : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5ad53bef0734b77a0861e332b82f24d : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/213e0176bf624713ad9ddc87b71aa653 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f0151e490c9e410b82137104cd875aca : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ee4b1cf6d5647a5848f34e251832a02 : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e9209aba5a3343e9b9b6fff88dcd6140 : SUCCESS in 12m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/348826a39afc45598ce0eed05830606c : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d1e0ca190e640c1834647fe88dc058f : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/317c2ef42eca44a18f2b823753eee9bf : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bbc646215fcb43c8a23dc72590572e4b : SUCCESS in 11m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/815080d7709246e69635951a943eb1b6 : SUCCESS in 13m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/18f4096d5a2a43e7bc456fddacc9691e : SUCCESS in 5m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8da39e58a48646bfa15f30b9e46cf829 : SUCCESS in 28m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/298866b2834c463e94aa49cea9247428 : SUCCESS in 12m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63fe5f196d304ca29c841830e24be1bf : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d11c187b46a748438c75a60a2e474b18 : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4a4b968e53b4875a9baf75bdb950c99 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf5de5d364e64a829e5198f56e315f86 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aca0bf6f00d4494e89e9e757c5b39930 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97122d8b499842dc8e5cee0c16cba2fc : SUCCESS in 48m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7f5bab2422b4a8f8bbf52ae24cce32c : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"72be0c44926c0a5fb0532b4dadb8ae81795fcee9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-15 18:45:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0dcd47909b06b07c38fc053cdfb40e89400e3e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:45:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1bb1518c350204e9702a7feb3b28c7c61754a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 20:32:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab688da55af41f3973c893d6282f478 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1699593be113467f98d40a7a38257fcd : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93e9172e6f34427093fe4f2d75decced : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fe8a864e0db42b9b2dec639625f1f82 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e706ee969a7540409f505eaede550b53 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d33383784df0419084bfd872ebf78df0 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7af38fa44bd46bebe4720e6f657206a : SUCCESS in 14m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/62de32dd22cd43519eca762080b73d1c : SUCCESS in 11m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f2794e12d4394c3faf49a11110fc634b : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f928dd7c082f489786dbd76251a44509 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a68b433a5316408c8907fc905aab5e7f : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66d3e4ef7a034b3eb2854582a8036b8b : SUCCESS in 11m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3c3fac83f974918b56b2641a5b0140e : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a326e5a57c74fbcaf634f781b20f217 : SUCCESS in 15m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/294f04f67128477dbfe290ed149fb8c2 : SUCCESS in 5m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/de84bf676dae46e6b0280545f6dc12c2 : SUCCESS in 33m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d21b5517d1048d5a3da48dbf31545f3 : SUCCESS in 14m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b40a01374b224f88a51db83bb1a6ad99 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1b828a83a5d452d954eedef49bf326f : SUCCESS in 20m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06fe999682044a6286b6aefb04060bc6 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a38d3e346aaa413d94fb9bb233b9f9ee : SUCCESS in 7m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f471efdc1f3c4af4bde9467328f93b66 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7319432bf154836a084f064f74cb890 : SUCCESS in 47m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/270828e2b6c24bc09f0f7869e354502a : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":2},{"id":"345cadb632644dbae1a59e34acc577b232bf3aff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 20:32:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fed932455caddb3d39072f6a611db90517e692a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 20:34:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9204c8aaa1514262b043ea5fa67dea5a : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3539e11b97fba2f8d94a451799bea52f4ba601ac","revisions":{"12303f491a5430cac3be48fc2ece10e329d1d255":{"kind":"REWORK","_number":1,"created":"2021-04-15 06:35:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/786360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/786360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/786360/1"}}},"commit":{"parents":[{"commit":"bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-15 06:35:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-15 06:35:23.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1da74bbd42defa7e2b22a9799fdc48ae4cdafa8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12303f491a5430cac3be48fc2ece10e329d1d255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12303f491a5430cac3be48fc2ece10e329d1d255"}]},"branch":"refs/heads/master"},"3539e11b97fba2f8d94a451799bea52f4ba601ac":{"kind":"REWORK","_number":2,"created":"2021-04-15 17:18:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/786360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/786360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/786360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/786360/2"}}},"commit":{"parents":[{"commit":"bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfeeee225ddddefc502c01e0fe7ef5bd68beaf5d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-15 06:35:23.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-15 16:04:46.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1da74bbd42defa7e2b22a9799fdc48ae4cdafa8b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3539e11b97fba2f8d94a451799bea52f4ba601ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3539e11b97fba2f8d94a451799bea52f4ba601ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
